Skip to content

NO-JIRA: Simplified kubectl component version check - #1662

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:kubectl-multiple-version
Jan 24, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:kubectl-multiple-version

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

With this PR #1656, we relaxed the kubectl component version
check by only allowing kubectl version in cli and cli-artifacts images and kubectl version between
other images. However, as seen in this revert #1660, there are other oc commands
having this strict component version check which caused failures in HCP.

This PR adopts simpler approach for kubectl specific component version checking. It simply
populates all the kubectl versions in a set and if the length is greater than 2(actual kubectl version 1.29.0 and the current ones 1.28.2) which means that this is indeed kubectl version mismatch.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 24, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@ardaguclu: This pull request explicitly references no jira issue.

Details

In response to this:

With this PR #1656, we relaxed the kubectl component version
check by only allowing kubectl version in cli and cli-artifacts images and kubectl version between
other images. However, as seen in this revert #1660, there are other oc commands
having this strict component version check which caused failures in HCP.

This PR adopts simpler approach for kubectl specific component version checking. It simply
populates all the kubectl versions in a set and if the length is greater than 2(actual kubectl version 1.29.0 and the current ones 1.28.2) which means that this is indeed kubectl version mismatch.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from mfojtik and soltysh January 24, 2024 07:21
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 24, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

/hold
until it is proved that this change works

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 24, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

For more justification: #1661 (comment)

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jan 24, 2024
@openshift-ci

openshift-ci Bot commented Jan 24, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jan 24, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@ardaguclu

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 24, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit 36702f9 into openshift:master Jan 24, 2024
@ardaguclu
ardaguclu deleted the kubectl-multiple-version branch January 24, 2024 11:10
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202401241315.p0.g36702f9.assembly.stream for distgit openshift-enterprise-cli.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants