Skip to content

WRKLDS-1016: Bump kube dependencies to 1.29.0 - #1661

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:revert-1660-revert-1655-bump-1.29.0
Jan 29, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:revert-1660-revert-1655-bump-1.29.0

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

This PR bumps k8s dependencies to 1.29.0 for OCP 4.16.

@openshift-ci-robot

openshift-ci-robot commented Jan 24, 2024 •

Copy link
Copy Markdown

@ardaguclu: This pull request references WRKLDS-1016 which is a valid jira issue.

Details

In response to this:

This PR bumps k8s dependencies to 1.29.0 for OCP 4.16.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 24, 2024
@openshift-ci
openshift-ci Bot requested review from deads2k and soltysh January 24, 2024 07:04
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 24, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 24, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

/test all

@ingvagabund

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jan 25, 2024
@openshift-ci

openshift-ci Bot commented Jan 25, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, ingvagabund

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ardaguclu

Copy link
Copy Markdown
Member Author

/test all

@ardaguclu

Copy link
Copy Markdown
Member Author

All the prerequisite PRs have merged;
/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 29, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

2 similar comments
@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 5d7c8d5 and 2 for PR HEAD 48b63d0 in total

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

3 similar comments
@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/tide help

@ardaguclu

Copy link
Copy Markdown
Member Author

/tide ?

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci

openshift-ci Bot commented Jan 29, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 6d67575 into openshift:master Jan 29, 2024
@ardaguclu
ardaguclu deleted the revert-1660-revert-1655-bump-1.29.0 branch January 29, 2024 20:16
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202401300919.p0.g6d67575.assembly.stream for distgit openshift-enterprise-cli.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants