Skip to content

Revert "WRKLDS-1016: Bump kube dependencies to 1.29.0" - #1660

Merged
openshift-merge-bot[bot] merged 1 commit into
masterfrom
revert-1655-bump-1.29.0
Jan 23, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
masterfrom
revert-1655-bump-1.29.0

Conversation

@sjenning

Copy link
Copy Markdown
Contributor

Reverts #1655

Per TRT policy, this broke CI release stream testing and is being reverted

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 23, 2024
@openshift-ci-robot

openshift-ci-robot commented Jan 23, 2024 •

Copy link
Copy Markdown

@sjenning: This pull request references WRKLDS-1016 which is a valid jira issue.

Details

In response to this:

Reverts #1655

Per TRT policy, this broke CI release stream testing and is being reverted

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@ardaguclu

Copy link
Copy Markdown
Member

/lgtm

@sjenning

Copy link
Copy Markdown
Contributor Author

/payload-job periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn

@sjenning

Copy link
Copy Markdown
Contributor Author

/hold

for payload test

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 23, 2024
@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

@sjenning: trigger 1 job(s) for the /payload-(job|aggregate|job-with-prs) command

  • periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/c7fc7c40-ba10-11ee-9268-9f84d59b8886-0

@ardaguclu

Copy link
Copy Markdown
Member

/approve
in case HCP tests passes with the revert.

@stbenjam

stbenjam commented Jan 23, 2024 •

Copy link
Copy Markdown
Member

No need for a payload test in this case, the cause looks pretty clear

/skip
/label approved
/lgtm

@openshift-ci openshift-ci Bot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Jan 23, 2024
@openshift-ci
openshift-ci Bot requested review from deads2k and mfojtik January 23, 2024 17:06
@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, sjenning, stbenjam

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@stbenjam

stbenjam commented Jan 23, 2024 •

Copy link
Copy Markdown
Member

/close

CI jobs seem stuck (github webhook issues)

@stbenjam

Copy link
Copy Markdown
Member

/reopen

@openshift-ci openshift-ci Bot closed this Jan 23, 2024
@stbenjam

Copy link
Copy Markdown
Member

/reopen

1 similar comment
@stbenjam

Copy link
Copy Markdown
Member

/reopen

@openshift-ci openshift-ci Bot reopened this Jan 23, 2024
@openshift-ci-robot

openshift-ci-robot commented Jan 23, 2024 •

Copy link
Copy Markdown

@sjenning: This pull request references WRKLDS-1016 which is a valid jira issue.

Details

In response to this:

Reverts #1655

Per TRT policy, this broke CI release stream testing and is being reverted

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@sjenning

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 23, 2024
@stbenjam

Copy link
Copy Markdown
Member

/override ci/prow/e2e-aws-ovn
/override ci/prow/e2e-agnostic-ovn-cmd
/override ci/prow/build-rpms-from-tar
/override ci/prow/security
/override ci/prow/rpm-build
/override ci/prow/e2e-metal-ipi-ovn-ipv6
/override ci/prow/e2e-aws-ovn-upgrade
/override ci/prow/e2e-aws-ovn-serial

@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

@stbenjam: /override requires failed status contexts, check run or a prowjob name to operate on.
The following unknown contexts/checkruns were given:

  • ci/prow/security

Only the following failed contexts/checkruns were expected:

  • ci/prow/build-rpms-from-tar
  • ci/prow/e2e-agnostic-ovn-cmd
  • ci/prow/e2e-aws-ovn
  • ci/prow/e2e-aws-ovn-serial
  • ci/prow/e2e-aws-ovn-upgrade
  • ci/prow/e2e-metal-ipi-ovn-ipv6
  • ci/prow/images
  • ci/prow/rpm-build
  • ci/prow/unit
  • ci/prow/verify
  • ci/prow/verify-deps
  • pull-ci-openshift-oc-master-build-rpms-from-tar
  • pull-ci-openshift-oc-master-e2e-agnostic-ovn-cmd
  • pull-ci-openshift-oc-master-e2e-aws-ovn
  • pull-ci-openshift-oc-master-e2e-aws-ovn-serial
  • pull-ci-openshift-oc-master-e2e-aws-ovn-upgrade
  • pull-ci-openshift-oc-master-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-oc-master-images
  • tide

If you are trying to override a checkrun that has a space in it, you must put a double quote on the context.

Details

In response to this:

/override ci/prow/e2e-aws-ovn
/override ci/prow/e2e-agnostic-ovn-cmd
/override ci/prow/build-rpms-from-tar
/override ci/prow/security
/override ci/prow/rpm-build
/override ci/prow/e2e-metal-ipi-ovn-ipv6
/override ci/prow/e2e-aws-ovn-upgrade
/override ci/prow/e2e-aws-ovn-serial

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 329f4d8 and 2 for PR HEAD d0d89a8 in total

@stbenjam

Copy link
Copy Markdown
Member

/override ci/prow/e2e-aws-ovn
/override ci/prow/e2e-agnostic-ovn-cmd
/override ci/prow/build-rpms-from-tar
/override ci/prow/security
/override ci/prow/rpm-build
/override ci/prow/e2e-metal-ipi-ovn-ipv6
/override ci/prow/e2e-aws-ovn-upgrade
/override ci/prow/e2e-aws-ovn-serial
/skip

@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

@stbenjam: /override requires failed status contexts, check run or a prowjob name to operate on.
The following unknown contexts/checkruns were given:

  • ci/prow/security

Only the following failed contexts/checkruns were expected:

  • ci/prow/build-rpms-from-tar
  • ci/prow/e2e-agnostic-ovn-cmd
  • ci/prow/e2e-aws-ovn
  • ci/prow/e2e-aws-ovn-serial
  • ci/prow/e2e-aws-ovn-upgrade
  • ci/prow/e2e-metal-ipi-ovn-ipv6
  • ci/prow/images
  • ci/prow/rpm-build
  • ci/prow/unit
  • ci/prow/verify
  • ci/prow/verify-deps
  • pull-ci-openshift-oc-master-build-rpms-from-tar
  • pull-ci-openshift-oc-master-e2e-agnostic-ovn-cmd
  • pull-ci-openshift-oc-master-e2e-aws-ovn
  • pull-ci-openshift-oc-master-e2e-aws-ovn-serial
  • pull-ci-openshift-oc-master-e2e-aws-ovn-upgrade
  • pull-ci-openshift-oc-master-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-oc-master-images
  • tide

If you are trying to override a checkrun that has a space in it, you must put a double quote on the context.

Details

In response to this:

/override ci/prow/e2e-aws-ovn
/override ci/prow/e2e-agnostic-ovn-cmd
/override ci/prow/build-rpms-from-tar
/override ci/prow/security
/override ci/prow/rpm-build
/override ci/prow/e2e-metal-ipi-ovn-ipv6
/override ci/prow/e2e-aws-ovn-upgrade
/override ci/prow/e2e-aws-ovn-serial
/skip

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@stbenjam

Copy link
Copy Markdown
Member

/override ci/prow/e2e-aws-ovn-serial
/override ci/prow/e2e-aws-ovn-upgrade
/override ci/prow/e2e-aws-ovn

@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

@stbenjam: Overrode contexts on behalf of stbenjam: ci/prow/e2e-aws-ovn, ci/prow/e2e-aws-ovn-serial, ci/prow/e2e-aws-ovn-upgrade

Details

In response to this:

/override ci/prow/e2e-aws-ovn-serial
/override ci/prow/e2e-aws-ovn-upgrade
/override ci/prow/e2e-aws-ovn

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants