Skip to content

NO-JIRA: Consider kubectl version checks only from cli and cli-artifact images - #1656

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:kubectl-cli-image-check
Jan 22, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:kubectl-cli-image-check

Conversation

@ardaguclu

@ardaguclu ardaguclu commented Jan 19, 2024 •

Copy link
Copy Markdown
Member

Currently, when we bump oc to newer k8s versions, our CI jobs are failing due to kubectl version mismatch error. The reason is that when we bump oc to newer kubectl(i.e. 1.29.0), cli and cli-artifact images manifest this version. On the other hand, other images in other operators have not-bumped kubectl version(i.e. 1.28.2).

This PR only populates kubectl versions in cli and cli-artifact images(because they are the source of truth) but in addition to that this PR checks the other images as well to have a consistency of a rule that all images should match either kubectl version in cli, cli-artifact or match with each other (all versions in other images at least should equal to 1.28.2).

valid image sets
cli 1.29.0
cli-artifact 1.29.0
others all 1.28.2
cli 1.29.0
cli-artifact 1.29.0
others all 1.29.0
cli 1.29.0
cli-artifact 1.29.0
others 1.28.2
another one  1.29.0
invalid image sets
cli 1.29.0
cli-artifact 1.29.0
others 1.28.2
another one 1.27.3
cli 1.29.0
cli-artifact 1.28.2
others all 1.28.2

This PR is prerequisite of #1655

Currently, when we bump oc to newer k8s versions, our CI jobs are
failing due to kubectl version mismatch error. The reason is that when we bump
oc to newer kubectl(i.e. 1.29.0), cli and cli-artifact images
manifest this version. On the other hand, other images in other operators
have not-bumped kubectl version(i.e. 1.28.2).

This PR only populates kubectl versions in cli and cli-artifact
images(because they are the source of truth) but in addition to that
this PR checks the other images as well to have a consistency with
a rule about all images should match either kubectl version in cli,
cli-artifact or match with each other (all versions in other images
at least should equal to 1.28.2).
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 19, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@ardaguclu: This pull request explicitly references no jira issue.

Details

In response to this:

Currently, when we bump oc to newer k8s versions, our CI jobs are failing due to kubectl version mismatch error. The reason is that when we bump oc to newer kubectl(i.e. 1.29.0), cli and cli-artifact images manifest this version. On the other hand, other images in other operators have not-bumped kubectl version(i.e. 1.28.2).

This PR only populates kubectl versions in cli and cli-artifact images(because they are the source of truth) but in addition to that this PR checks the other images as well to have a consistency with a rule about all images should match either kubectl version in cli, cli-artifact or match with each other (all versions in other images at least should equal to 1.28.2).

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@ardaguclu

Copy link
Copy Markdown
Member Author

/hold
until we prove that this patch works.

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 19, 2024
@openshift-ci
openshift-ci Bot requested review from mfojtik and soltysh January 19, 2024 11:42
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 19, 2024
@ardaguclu ardaguclu changed the title NO-JIRA: Consider kubectl version checks only in cli and cli-artifact images NO-JIRA: Consider kubectl version checks only from cli and cli-artifact images Jan 19, 2024
@ardaguclu

ardaguclu commented Jan 19, 2024 •

Copy link
Copy Markdown
Member Author

Spun up a cluster that contains a combination of this PR and #1656 and all tests passed;

https://prow.ci.openshift.org/view/gs/test-platform-results/logs/release-openshift-origin-installer-launch-aws-modern/1748310181223075840

@ardaguclu

Copy link
Copy Markdown
Member Author

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 19, 2024

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jan 22, 2024
@openshift-ci

openshift-ci Bot commented Jan 22, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci

openshift-ci Bot commented Jan 22, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 64c6784 into openshift:master Jan 22, 2024
@ardaguclu
ardaguclu deleted the kubectl-cli-image-check branch January 22, 2024 13:57
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202401230706.p0.g64c6784.assembly.stream for distgit openshift-enterprise-cli.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants