Skip to content

OCPBUGS-56278: extensions: llvm now available in rhel 9.6 and 10 - #1817

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:release-4.19from
jbtrystram:llvm-9.6
May 16, 2025
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:release-4.19from
jbtrystram:llvm-9.6

Conversation

@jbtrystram

Copy link
Copy Markdown
Contributor

Closes #1680

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 12, 2025
@jbtrystram jbtrystram changed the title extensions: llvm now available in rhel 9.6 and 10 COS-3075: extensions: llvm now available in rhel 9.6 and 10 May 12, 2025
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label May 12, 2025
@openshift-ci-robot

openshift-ci-robot commented May 12, 2025 •

Copy link
Copy Markdown

@jbtrystram: This pull request references COS-3075 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the task to target either version "4.20." or "openshift-4.20.", but it targets "4.19" instead.

Details

In response to this:

Closes #1680

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jbtrystram

Copy link
Copy Markdown
Contributor Author

/jira-refresh

@mike-nguyen

Copy link
Copy Markdown
Member

Does the crun-wasm extension get built without failing now?

@travier travier left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/hold
(until we've confirmed this fixes the issue :))

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 12, 2025
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 12, 2025
@jbtrystram

Copy link
Copy Markdown
Contributor Author

Yeah some llvm libs are still missing :(

Importing rpm-md...done
rpm-md repo 'rhel-9.6-baseos'; generated: 2025-04-16T23:25:48Z solvables: 1173
rpm-md repo 'rhel-9.6-appstream'; generated: 2025-04-16T23:25:27Z solvables: 6262
rpm-md repo 'rhel-9.6-fast-datapath'; generated: 2025-04-17T00:12:31Z solvables: 1201
rpm-md repo 'rhel-9.6-nfv'; generated: 2025-04-16T23:26:27Z solvables: 21
rpm-md repo 'rhel-9.6-highavailability'; generated: 2025-04-16T23:26:23Z solvables: 44
rpm-md repo 'rhel-9.6-server-ose-4.19'; generated: 2025-05-12T10:24:26Z solvables: 526
rpm-md repo 'rhel-9.6-early-kernel'; generated: 2025-05-12T10:24:26Z solvables: 526
Resolving dependencies...done
error: Could not depsolve transaction; 1 problem detected:
 Problem: conflicting requests
  - package crun-wasm-1.21-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19 requires crun = 1.21-2.rhaos4.19.el9, but none of the providers can be installed
  - package crun-wasm-1.21-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel requires crun = 1.21-2.rhaos4.19.el9, but none of the providers can be installed
  - cannot install both crun-1.21-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19 and crun-1.19.1-1.el9.x86_64 from @System
  - cannot install both crun-1.21-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel and crun-1.19.1-1.el9.x86_64 from @System
  - package crun-wasm-1.8.5-4.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19 requires wasm-library, but none of the providers can be installed
  - package crun-wasm-1.8.5-5.rhaos4.14.el9.x86_64 from rhel-9.6-server-ose-4.19 requires wasm-library, but none of the providers can be installed
  - package crun-wasm-1.8.5-4.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel requires wasm-library, but none of the providers can be installed
  - package crun-wasm-1.8.5-5.rhaos4.14.el9.x86_64 from rhel-9.6-early-kernel requires wasm-library, but none of the providers can be installed
  - nothing provides libLLVM-17.so()(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides libLLVM-17.so(LLVM_17)(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides liblldCommon.so.17()(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides liblldELF.so.17()(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides libLLVM-17.so()(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides libLLVM-17.so(LLVM_17)(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides liblldCommon.so.17()(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides liblldELF.so.17()(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-server-ose-4.19
  - nothing provides libLLVM-17.so()(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides libLLVM-17.so(LLVM_17)(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides liblldCommon.so.17()(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides liblldELF.so.17()(64bit) needed by wasmedge-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides libLLVM-17.so()(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides libLLVM-17.so(LLVM_17)(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides liblldCommon.so.17()(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel
  - nothing provides liblldELF.so.17()(64bit) needed by wasmedge-rt-0.14.0-2.rhaos4.19.el9.x86_64 from rhel-9.6-early-kernel

@jbtrystram

jbtrystram commented May 12, 2025 •

Copy link
Copy Markdown
Contributor Author

Does the crun-wasm extension get built without failing now?

no it does not. I thought the missing package was llvm and was asked by ART why we were pulling an old version while a newer exist.
I misunderstood the underlying issue.

@dustymabe

Copy link
Copy Markdown
Member

also related to this: #1680 (comment)

@travier

travier commented May 12, 2025

Copy link
Copy Markdown
Member

/lgtm cancel

@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label May 12, 2025
@joelcapitao

Copy link
Copy Markdown
Contributor

/lgtm cancel

@jlebon

jlebon commented May 12, 2025

Copy link
Copy Markdown
Member

I think the LLVM errors there are a red herring and this is really just #1680 (comment).

@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

@jbtrystram

jbtrystram commented May 14, 2025 •

Copy link
Copy Markdown
Contributor Author

I had this passing fine in the pipeline so this has to do with how the repos are mirrored in PROW. let's see how it goes after openshift/release#64914 merges

@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented May 15, 2025

Copy link
Copy Markdown
Contributor

@jbtrystram: The /retest command does not accept any targets.
The following commands are available to trigger required jobs:

/test images

The following commands are available to trigger optional jobs:

/test e2e-aws
/test okd-scos-e2e-aws-ovn
/test okd-scos-images

Use /test all to run the following jobs that were automatically triggered:

pull-ci-openshift-os-master-e2e-aws
pull-ci-openshift-os-master-images
pull-ci-openshift-os-master-okd-scos-e2e-aws-ovn
Details

In response to this:

/retest images

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

@jbtrystram
jbtrystram changed the base branch from master to release-4.19 May 15, 2025 17:23
@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@jbtrystram

Copy link
Copy Markdown
Contributor Author

/retest

@jbtrystram

Copy link
Copy Markdown
Contributor Author

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 15, 2025
@mike-nguyen

Copy link
Copy Markdown
Member

/label backport-risk-assessed
/label cherry-pick-approved

@openshift-ci openshift-ci Bot added backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. labels May 15, 2025
@jlebon

jlebon commented May 15, 2025

Copy link
Copy Markdown
Member

Side note: openshift/release#64590 landed after branching I think but it'd make sense to apply the changes to ci-operator/config/openshift/os/openshift-os-master.yaml in that PR also to the 4.19 version so that these jobs stop running on this branch.

@mike-nguyen

Copy link
Copy Markdown
Member

/retitle OCPBUGS-56278: extensions: llvm now available in rhel 9.6 and 10

@openshift-ci openshift-ci Bot changed the title COS-3075: extensions: llvm now available in rhel 9.6 and 10 OCPBUGS-56278: extensions: llvm now available in rhel 9.6 and 10 May 15, 2025
@openshift-ci-robot openshift-ci-robot added jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels May 15, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@jbtrystram: This pull request references Jira Issue OCPBUGS-56278, which is invalid:

  • release note text must be set and not match the template OR release note type must be set to "Release Note Not Required". For more information you can reference the OpenShift Bug Process.

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Closes #1680

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@mike-nguyen

Copy link
Copy Markdown
Member

/jira refresh

@openshift-ci-robot

Copy link
Copy Markdown

@mike-nguyen: This pull request references Jira Issue OCPBUGS-56278, which is invalid:

  • release note text must be set and not match the template OR release note type must be set to "Release Note Not Required". For more information you can reference the OpenShift Bug Process.

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@mike-nguyen

Copy link
Copy Markdown
Member

/jira refresh

@openshift-ci-robot openshift-ci-robot added jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. and removed jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels May 15, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@mike-nguyen: This pull request references Jira Issue OCPBUGS-56278, which is valid. The bug has been moved to the POST state.

7 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.19.0) matches configured target version for branch (4.19.0)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, POST)
  • release note type set to "Release Note Not Required"
  • dependent bug Jira Issue OCPBUGS-56277 is in the state ON_QA, which is one of the valid states (MODIFIED, ON_QA, VERIFIED)
  • dependent Jira Issue OCPBUGS-56277 targets the "4.20.0" version, which is one of the valid target versions: 4.20.0
  • bug has dependents

Requesting review from QA contact:
/cc @mike-nguyen

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested a review from mike-nguyen May 15, 2025 21:00
@openshift-ci

openshift-ci Bot commented May 15, 2025

Copy link
Copy Markdown
Contributor

@jbtrystram: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws ab142c9 link false /test e2e-aws
ci/prow/images ab142c9 link true /test images

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@mike-nguyen

Copy link
Copy Markdown
Member

/retest

@jlebon

jlebon commented May 16, 2025

Copy link
Copy Markdown
Member

Nice!

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 16, 2025
@openshift-ci

openshift-ci Bot commented May 16, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jbtrystram, jlebon

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 2acbcb0 into openshift:release-4.19 May 16, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@jbtrystram: Jira Issue OCPBUGS-56278: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-56278 has been moved to the MODIFIED state.

Details

In response to this:

Closes #1680

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[4.19-9.6] crun-wasm extension fails to build due dependency issues

8 participants