Skip to content

NO-JIRA: bump verison to 4.20 - #1804

Merged
openshift-merge-bot[bot] merged 3 commits into
openshift:masterfrom
mike-nguyen:420_manifest
May 15, 2025
Merged

openshift-merge-bot[bot] merged 3 commits into
openshift:masterfrom
mike-nguyen:420_manifest

Conversation

@mike-nguyen

Copy link
Copy Markdown
Member

No description provided.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 17, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@mike-nguyen: This pull request explicitly references no jira issue.

Details

In response to this:

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@mike-nguyen

Copy link
Copy Markdown
Member Author

/hold
until branch day

@openshift-ci
openshift-ci Bot requested review from cgwalters and jlebon April 17, 2025 21:18
@openshift-ci openshift-ci Bot added do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Apr 17, 2025
@mike-nguyen

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 14, 2025
@jbtrystram

Copy link
Copy Markdown
Contributor

/retest

The RHEL 9.6 repos have older version of packages now and it is
possible to build the old crun-wasm with the older llvm package.
@mike-nguyen

Copy link
Copy Markdown
Member Author

@jbtrystram I rolled in your changes from #1817 into this PR to fix CI but didn't rebase on top of your PR. Let me know if you want me to rebase on top of your PR as it was work you did.

@mike-nguyen

Copy link
Copy Markdown
Member Author

/test images

@jbtrystram

Copy link
Copy Markdown
Contributor

/restest

@mike-nguyen

Copy link
Copy Markdown
Member Author

/retest

@jbtrystram

Copy link
Copy Markdown
Contributor

@jbtrystram I rolled in your changes from #1817 into this PR to fix CI but didn't rebase on top of your PR. Let me know if you want me to rebase on top of your PR as it was work you did.

No worries. I'll rebase my own on the 4.19 branch

@openshift-ci

openshift-ci Bot commented May 15, 2025

Copy link
Copy Markdown
Contributor

@mike-nguyen: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@mike-nguyen

Copy link
Copy Markdown
Member Author

Tests passed!

@jbtrystram

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 15, 2025
@openshift-ci

openshift-ci Bot commented May 15, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jbtrystram, mike-nguyen

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [jbtrystram,mike-nguyen]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 4f1036b into openshift:master May 15, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants