Skip to content

config: add coreos/rhel-coreos-config - #64590

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
jlebon:pr/rhcos-split
May 7, 2025
Merged

openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
jlebon:pr/rhcos-split

Conversation

@jlebon

@jlebon jlebon commented May 6, 2025

Copy link
Copy Markdown
Member

This is a new repo that was split out of openshift/os. Add configuration for it.

Move all the cosa-based tests from openshift/os to this new repo.

@openshift-ci
openshift-ci Bot requested review from jmarrero and jmguzik May 6, 2025 20:03
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 6, 2025
@jlebon
jlebon force-pushed the pr/rhcos-split branch 4 times, most recently from 1f6ac62 to 4173a81 Compare May 7, 2025 10:56
jlebon added 2 commits May 7, 2025 07:11
This is a new repo that was split out of openshift/os. Add configuration
for it.

Move all the cosa-based tests from openshift/os to this new repo.
This will be synced from the repo itself by automation in the future.
But to get this repo enrolled, AIUI we need to seed this info manually
to get CI to pass.
@jlebon
jlebon force-pushed the pr/rhcos-split branch from 4173a81 to fff53d0 Compare May 7, 2025 11:12
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@jlebon: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-coreos-rhel-coreos-config-main-images coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-rhcos-9-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-rhcos-9-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-rhcos-9next-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-rhcos-9next-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-scos-10-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-scos-10-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-scos-9-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-scos-9-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-main-validate coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-images coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-rhcos-9-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-rhcos-9-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-rhcos-9next-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-rhcos-9next-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-scos-10-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-scos-10-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-scos-9-build-test-metal coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-scos-9-build-test-qemu coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-coreos-rhel-coreos-config-rhel-coreos-config-main-validate coreos/rhel-coreos-config presubmit Presubmit changed
pull-ci-openshift-os-master-e2e-aws openshift/os presubmit Ci-operator config changed
pull-ci-openshift-os-master-images openshift/os presubmit Ci-operator config changed
pull-ci-openshift-os-master-okd-scos-e2e-aws-ovn openshift/os presubmit Ci-operator config changed
pull-ci-openshift-os-master-okd-scos-images openshift/os presubmit Ci-operator config changed
Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@jlebon

jlebon commented May 7, 2025

Copy link
Copy Markdown
Member Author

/pj-rehearse pull-ci-coreos-rhel-coreos-config-main-images pull-ci-openshift-os-master-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jlebon: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@jlebon

jlebon commented May 7, 2025

Copy link
Copy Markdown
Member Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jlebon: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label May 7, 2025
@jbtrystram

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 7, 2025

@prestist prestist left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Lgtm

@openshift-ci

openshift-ci Bot commented May 7, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jbtrystram, jlebon, prestist

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented May 7, 2025

Copy link
Copy Markdown
Contributor

@jlebon: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 85341e7 into openshift:master May 7, 2025
@openshift-ci

openshift-ci Bot commented May 7, 2025

Copy link
Copy Markdown
Contributor

@jlebon: Updated the following 2 configmaps:

  • plugins configmap in namespace ci at cluster app.ci using the following files:
    • key core-services-prow-02_config-coreos-rhel-coreos-config-_pluginconfig.yaml using file core-services/prow/02_config/coreos/rhel-coreos-config/_pluginconfig.yaml
  • config configmap in namespace ci at cluster app.ci using the following files:
    • key core-services-prow-02_config-coreos-rhel-coreos-config-_prowconfig.yaml using file core-services/prow/02_config/coreos/rhel-coreos-config/_prowconfig.yaml
Details

In response to this:

This is a new repo that was split out of openshift/os. Add configuration for it.

Move all the cosa-based tests from openshift/os to this new repo.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

tbuskey pushed a commit to tbuskey/release that referenced this pull request May 19, 2025
* config: add coreos/rhel-coreos-config

This is a new repo that was split out of openshift/os. Add configuration
for it.

Move all the cosa-based tests from openshift/os to this new repo.

* config: add OWNERS file for coreos/rhel-coreos-config

This will be synced from the repo itself by automation in the future.
But to get this repo enrolled, AIUI we need to seed this info manually
to get CI to pass.
tareqalayan pushed a commit to tareqalayan/release that referenced this pull request May 26, 2025
* config: add coreos/rhel-coreos-config

This is a new repo that was split out of openshift/os. Add configuration
for it.

Move all the cosa-based tests from openshift/os to this new repo.

* config: add OWNERS file for coreos/rhel-coreos-config

This will be synced from the repo itself by automation in the future.
But to get this repo enrolled, AIUI we need to seed this info manually
to get CI to pass.
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants