Skip to content

WRKLDS-1016: Bump kube dependencies to 1.29.0 - #1655

Merged
openshift-merge-bot[bot] merged 7 commits into
openshift:masterfrom
ardaguclu:bump-1.29.0
Jan 23, 2024
Merged

openshift-merge-bot[bot] merged 7 commits into
openshift:masterfrom
ardaguclu:bump-1.29.0

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

This PR bumps k8s dependencies to 1.29.0 for OCP 4.16.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 17, 2024
@openshift-ci-robot

openshift-ci-robot commented Jan 17, 2024 •

Copy link
Copy Markdown

@ardaguclu: This pull request references WRKLDS-1016 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the epic to target either version "4.16." or "openshift-4.16.", but it targets "openshift-4.15" instead.

Details

In response to this:

This PR bumps k8s dependencies to 1.29.0 for OCP 4.16.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from deads2k and mfojtik January 17, 2024 06:05
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 17, 2024
@ardaguclu ardaguclu changed the title WRKLDS-1016: Bump kube dependencies to 1.29.0 WIP: WRKLDS-1016: Bump kube dependencies to 1.29.0 Jan 17, 2024
@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jan 17, 2024
@ardaguclu ardaguclu changed the title WIP: WRKLDS-1016: Bump kube dependencies to 1.29.0 WRKLDS-1016: Bump kube dependencies to 1.29.0 Jan 17, 2024
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jan 17, 2024
@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jan 22, 2024
@openshift-merge-robot openshift-merge-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jan 22, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

Let's retest assuming that this #1656 is landed in images.
/retest

@ingvagabund

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jan 23, 2024
@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, ingvagabund

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jan 23, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 329f4d8 into openshift:master Jan 23, 2024
@ardaguclu
ardaguclu deleted the bump-1.29.0 branch January 23, 2024 09:28
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202401231834.p0.g329f4d8.assembly.stream for distgit openshift-enterprise-cli.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants