CONSOLE-5031: Remove Firehose Component and Fix and Cleanup Impacted Types - #16096
Conversation
|
@cajieh: This pull request references CONSOLE-5031 which is a valid jira issue. Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.22.0" version, but no target version was set. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
1680ec9 to
5b2a87b
Compare
5b2a87b to
51a81d3
Compare
78d187e to
4e718b8
Compare
4e718b8 to
3d41d3e
Compare
|
/retest |
jhadvig
left a comment
There was a problem hiding this comment.
👑 Thank you Cyril for making this happen.
... farewell to our old nemesis 🔥
/lgtm
| | Name | Value Type | Optional | Description | | ||
| | ---- | ---------- | -------- | ----------- | | ||
| | `k8sResource` | `CodeRef<FirehoseResource & { isList: true; }>` | no | The utilization item to be replaced. | | ||
| | `k8sResource` | `CodeRef<WatchK8sResource & { prop: string; } & { isList: true; }>` | no | The utilization item to be replaced. | |
There was a problem hiding this comment.
we need to update the 4.22 changelog with this breaking change
| dataSelector: string[] | number[] | symbol[]; | ||
| resources: WatchK8sResultsObject<K8sResourceKind | K8sResourceKind[]>[]; | ||
| showBadge?: boolean; | ||
| onLoad?: (items: ResourceDropdownItems) => void; | ||
| onChange?: ResourceDropdownProps['onChange']; | ||
| resourceFilter?: ResourceDropdownProps['resourceFilter']; | ||
| autoSelect?: ResourceDropdownProps['autoSelect']; | ||
| resourceFilter?: (resource: K8sResourceKind) => boolean; | ||
| autoSelect?: boolean; | ||
| placeholder?: string; | ||
| actionItems?: ResourceDropdownProps['actionItems']; | ||
| appendItems?: ResourceDropdownProps['appendItems']; | ||
| customResourceKey?: ResourceDropdownProps['customResourceKey']; | ||
| actionItems?: { | ||
| actionTitle: string; | ||
| actionKey: string; | ||
| }[]; | ||
| appendItems?: ResourceDropdownItems; | ||
| customResourceKey?: (key: string, resource: K8sResourceKind) => string; | ||
| dataTest?: string; | ||
| menuClassName?: ResourceDropdownProps['menuClassName']; | ||
| menuClassName?: string; |
| [props.namespace], | ||
| ); | ||
|
|
||
| const watchedResources = useK8sWatchResources<{ secrets: SecretKind[] }>(watchSpec); |
There was a problem hiding this comment.
nit, if we only need to watch secrets, why use useK8sWatchResources instead of useK8sWatchResource?
c8ba7bb to
2798f36
Compare
|
/assign @yapei |
|
/test e2e-gcp-console |
1 similar comment
|
/test e2e-gcp-console |
|
it looks like we have some issues loading Source secret on Import from Git page, a loading spinning icon is always showing and newly created secret are not shown in dropdown list for other changes, no regression issues found issues-loading-source-secret.mov |
2798f36 to
fad32d6
Compare
fad32d6 to
15b7e99
Compare
|
the reported issue has been fixed in latest code |
|
@yapei: This PR has been marked as verified by DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/test e2e-gcp-console |
|
@cajieh: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: cajieh, jhadvig, logonoff, TheRealJon The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
### NOT READY FOR REVIEWWaiting on PR to merge.Summary by CodeRabbit
Release Notes