Skip to content

hypershift: run e2e-v2 with tech preview - #76767

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
sjenning:e2e-v2-tp
Mar 24, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
sjenning:e2e-v2-tp

Conversation

@sjenning

@sjenning sjenning commented Mar 24, 2026

Copy link
Copy Markdown
Contributor

We have an issue with GCP API UX checks bit rotting in e2e-v2. This should fix that.

@openshift-ci
openshift-ci Bot requested review from enxebre and liangxia March 24, 2026 16:37
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 24, 2026
@enxebre

enxebre commented Mar 24, 2026

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Mar 24, 2026
@openshift-ci

openshift-ci Bot commented Mar 24, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: enxebre, sjenning

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-main-e2e-v2-aws

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@sjenning: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-hypershift-release-4.22-e2e-v2-aws openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.23-e2e-v2-aws openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-5.0-e2e-v2-aws openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-main-e2e-v2-aws openshift/hypershift presubmit Ci-operator config changed

Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Mar 24, 2026
@openshift-ci

openshift-ci Bot commented Mar 24, 2026

Copy link
Copy Markdown
Contributor

@sjenning: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/openshift/hypershift/main/e2e-v2-aws e3c7fd8 link unknown /pj-rehearse pull-ci-openshift-hypershift-main-e2e-v2-aws

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 9dcdec0 into openshift:main Mar 24, 2026
13 of 14 checks passed
xiaozhisean pushed a commit to xiaozhisean/openshift-release that referenced this pull request Mar 24, 2026
eliorerz pushed a commit to eliorerz/release that referenced this pull request Jun 3, 2026
fjglira pushed a commit to fjglira/release that referenced this pull request Jun 3, 2026
andrej1991 pushed a commit to andrej1991/release that referenced this pull request Jul 6, 2026
SachinNinganure pushed a commit to SachinNinganure/release that referenced this pull request Jul 20, 2026
malingatembo pushed a commit to malingatembo/release that referenced this pull request Aug 5, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants