Skip to content

remove tide configuration entirely for ARO-HCP - #66524

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
deads2k:remove-tide
Jun 27, 2025
Merged

remove tide configuration entirely for ARO-HCP#66524
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
deads2k:remove-tide

Conversation

@deads2k

@deads2k deads2k commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

possible alternative to #66523 trying to stop tide

@openshift-ci
openshift-ci Bot requested review from pruan-rht and smg247 June 27, 2025 13:53
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 27, 2025
@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

@deads2k: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jun 27, 2025
@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: deads2k, geoberle

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@deads2k deads2k added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Jun 27, 2025
@openshift-merge-bot
openshift-merge-bot Bot merged commit 3d1924e into openshift:master Jun 27, 2025
@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

@deads2k: Updated the config configmap in namespace ci at cluster app.ci using the following files:

  • key core-services-prow-02_config-Azure-ARO-HCP-_prowconfig.yaml using file core-services/prow/02_config/Azure/ARO-HCP/_prowconfig.yaml
Details

In response to this:

possible alternative to #66523 trying to stop tide

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

liweinan pushed a commit to liweinan/release that referenced this pull request Aug 5, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants