Skip to content

disable tide on ARO-HCP because we don't own merging - #66523

Closed
deads2k wants to merge 1 commit into
openshift:masterfrom
deads2k:disable-tide
Closed

disable tide on ARO-HCP because we don't own merging#66523
deads2k wants to merge 1 commit into
openshift:masterfrom
deads2k:disable-tide

Conversation

@deads2k

@deads2k deads2k commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: deads2k

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci
openshift-ci Bot requested review from jmguzik and pruan-rht June 27, 2025 13:48
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 27, 2025
@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Jun 27, 2025
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@deads2k: no rehearsable tests are affected by this change

@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

@deads2k: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/prow-config e1601bb link true /test prow-config
ci/prow/check-gh-automation-tide e1601bb link true /test check-gh-automation-tide
ci/prow/config e1601bb link true /test config

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@deads2k deads2k closed this Jun 27, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants