Skip to content

NO-JIRA: hypershift: add hypershift base_images to csi-operator config - #57317

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
sjenning:hypershift-csi-base-images
Oct 1, 2024
Merged

NO-JIRA: hypershift: add hypershift base_images to csi-operator config#57317
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
sjenning:hypershift-csi-base-images

Conversation

@sjenning

Copy link
Copy Markdown
Contributor

Fix misconfiguration uncovered by #57123

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: This pull request explicitly references no jira issue.

Details

In response to this:

Fix misconfiguration uncovered by #57123

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 30, 2024
@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-csi-operator-master-hypershift-aws-e2e-external

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@sjenning: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-csi-operator-release-4.19-aws-efs-operator-e2e openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-aws-efs-operator-e2e-extended openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-ci-index-aws-efs-csi-driver-operator-bundle openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-ci-index-smb-csi-driver-operator-bundle openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-aws-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-aws-csi-extended openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-aws-ovn-upgrade openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-csi-extended openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-file-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-file-csi-extended openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-file-nfs-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-manual-oidc openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azure-ovn-upgrade openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-azurestack-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-openstack openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-openstack-cinder-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-e2e-openstack-manila-csi openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-hypershift-aws-e2e-external openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-images openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-security openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-smb-operator-e2e openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-smb-operator-e2e-extended openshift/csi-operator presubmit Ci-operator config changed
pull-ci-openshift-csi-operator-release-4.19-unit openshift/csi-operator presubmit Ci-operator config changed

A total of 143 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@sjenning

Copy link
Copy Markdown
Contributor Author

e2e run flaked, but the job ran to completion
/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Sep 30, 2024
@sjenning

sjenning commented Oct 1, 2024

Copy link
Copy Markdown
Contributor Author

cc @jsafrane

@jsafrane

jsafrane commented Oct 1, 2024

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 1, 2024
@openshift-ci

openshift-ci Bot commented Oct 1, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jsafrane, sjenning

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 1, 2024
@openshift-ci

openshift-ci Bot commented Oct 1, 2024

Copy link
Copy Markdown
Contributor

@sjenning: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/openshift/csi-operator/master/hypershift-aws-e2e-external ae1a229 link unknown /pj-rehearse pull-ci-openshift-csi-operator-master-hypershift-aws-e2e-external

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 1aee42f into openshift:master Oct 1, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants