Skip to content

HOSTEDCP-1569: hypershift: use HO/e2e/cli from main on presubmits - #57123

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
sjenning:stop-publishing-branch-images
Sep 25, 2024
Merged

HOSTEDCP-1569: hypershift: use HO/e2e/cli from main on presubmits#57123
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
sjenning:stop-publishing-branch-images

Conversation

@sjenning

@sjenning sjenning commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

This switches all presubmits on the release-* branches 4.14+ of the hypershift repo to use HO/CLI/e2e from main

This also stops promoting the hypershift-operator and hypershift-tests images from the release branches 4.14+.

This does not impact MCE HO builds as they have their own postsubmits in the *__mce.yaml configs.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 25, 2024
@openshift-ci-robot

openshift-ci-robot commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

@sjenning: This pull request references HOSTEDCP-1569 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the epic to target the "4.18.0" version, but no target version was set.

Details

In response to this:

This switches all presubmits on the release-* branches of the hypershift repo to use HO/CLI/e2e from main

This also stops promoting the hypershift-operator and hypershift-tests images from the release branches.

This does not impact MCE HO builds as they have their own postsubmits in the *__mce.yaml configs.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from csrwng and enxebre September 25, 2024 02:22
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 25, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@sjenning: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-cluster-storage-operator-master-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.19-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.18-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.17-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.16-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.15-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.14-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.12-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-etcd-openshift-4.19-e2e-hypershift openshift/etcd presubmit Registry content changed
pull-ci-openshift-etcd-openshift-4.18-e2e-hypershift openshift/etcd presubmit Registry content changed
pull-ci-openshift-etcd-openshift-4.17-e2e-hypershift openshift/etcd presubmit Registry content changed
pull-ci-openshift-etcd-openshift-4.16-e2e-hypershift openshift/etcd presubmit Registry content changed
pull-ci-openshift-etcd-openshift-4.15-e2e-hypershift openshift/etcd presubmit Registry content changed
pull-ci-openshift-etcd-openshift-4.14-e2e-hypershift openshift/etcd presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-master-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-release-4.19-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-release-4.18-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-release-4.17-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-release-4.16-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-release-4.15-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-cluster-version-operator-release-4.14-e2e-hypershift openshift/cluster-version-operator presubmit Registry content changed
pull-ci-openshift-api-master-e2e-aws-ovn-hypershift openshift/api presubmit Registry content changed
pull-ci-openshift-api-release-4.19-e2e-aws-ovn-hypershift openshift/api presubmit Registry content changed
pull-ci-openshift-api-release-4.18-e2e-aws-ovn-hypershift openshift/api presubmit Registry content changed
pull-ci-openshift-api-release-4.17-e2e-aws-ovn-hypershift openshift/api presubmit Registry content changed

A total of 313 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@openshift-ci-robot

openshift-ci-robot commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

@sjenning: This pull request references HOSTEDCP-1569 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the epic to target the "4.18.0" version, but no target version was set.

Details

In response to this:

This switches all presubmits on the release-* branches 4.14+ of the hypershift repo to use HO/CLI/e2e from main

This also stops promoting the hypershift-operator and hypershift-tests images from the release branches 4.14+.

This does not impact MCE HO builds as they have their own postsubmits in the *__mce.yaml configs.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-main-e2e-aws

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-main-e2e-conformance

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-main-e2e-kubevirt-aws-ovn-reduced

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-release-4.14-e2e-aws

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-release-4.14-e2e-conformance

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-release-4.17-e2e-openstack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Sep 25, 2024
@openshift-ci

openshift-ci Bot commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

@sjenning: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@muraee

muraee commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 25, 2024
@openshift-ci

openshift-ci Bot commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: muraee, sjenning

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit d33ffdb into openshift:master Sep 25, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants