Skip to content

Find a source for google-cloud-sdk-183 - #8286

Merged
sdodson merged 1 commit into
openshift:masterfrom
sdodson:fix-gcp
May 7, 2018
Merged

Find a source for google-cloud-sdk-183#8286
sdodson merged 1 commit into
openshift:masterfrom
sdodson:fix-gcp

Conversation

@sdodson

@sdodson sdodson commented May 7, 2018

Copy link
Copy Markdown
Member

My least favorable fix to this problem but seems like the only thing that may work.

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sdodson

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels May 7, 2018
@sdodson

sdodson commented May 7, 2018

Copy link
Copy Markdown
Member Author

/test gcp

@openshift-ci-robot openshift-ci-robot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels May 7, 2018
@openshift-ci-robot openshift-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels May 7, 2018
@openshift-ci-robot openshift-ci-robot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels May 7, 2018
@sdodson sdodson changed the title Rebase to upstream gcp dynamic inventory Find a source for google-cloud-sdk-183 May 7, 2018
@sdodson
sdodson merged commit 1369205 into openshift:master May 7, 2018
@sdodson

sdodson commented May 8, 2018

Copy link
Copy Markdown
Member Author

/cherrypick release-3.9

@openshift-cherrypick-robot

Copy link
Copy Markdown

@sdodson: #8286 failed to apply on top of branch "release-3.9":

error: Failed to merge in the changes.
Using index info to reconstruct a base tree...
M	images/installer/Dockerfile
Falling back to patching base and 3-way merge...
Auto-merging images/installer/Dockerfile
CONFLICT (content): Merge conflict in images/installer/Dockerfile
Patch failed at 0001 fix

Details

In response to this:

/cherrypick release-3.9

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants