Forward porting of azure work to master - #8257
Conversation
|
|
||
| - name: setup systemd units | ||
| import_role: | ||
| name: openshift_master |
There was a problem hiding this comment.
This is not needed and will be removed. It is however used in the 3.9 branch. @jim-minter, should I remove this in this cherry-pick or should I add a follow on commit to remove, or should we clean up in subsequent PR since this will be the work required to bring up masters on 3.10?
This will update to roles/openshift_control_plane
There was a problem hiding this comment.
remove it in the cherry-pick in this PR. If there are additions in subsequent PRs that's fine
There was a problem hiding this comment.
yeah, this will have to be removed for tox to pass
|
This PR has the |
| sslclientkey: /var/lib/yum/client-key.pem | ||
| enabled: yes | ||
|
|
||
| - name: rhel-server-7-ose-3.9 |
There was a problem hiding this comment.
s/3.9/3.10 here and comment out the block?
|
|
||
| CentOS: | ||
| # TODO: should be using a repo which only provides prerequisites | ||
| - name: openshift-origin39 |
There was a problem hiding this comment.
s/39// this line and next
| - verify API server | ||
|
|
||
| - name: verify API server | ||
| - name: Verify API Server |
|
minor nits, otherwise looks fine to me. Once tests are passing I'm in favour of getting this in now and handling further changes required to the azure playbooks as follow-ups. |
316c18b to
af3d652
Compare
|
@jim-minter, cleaned up and removed |
|
@kwoodson lgtm! |
|
@sdodson, Ready when you are. This should introduce the owners file which will allow @jim-minter or myself to handle other PRs in the Azure playbooks area. Thanks |
|
/lgtm |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: kwoodson, sdodson The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Azure publishing portal requires input blob names to end in .vhd.
|
/lgtm |
|
/test gcp |
|
/test system-containers |
I'm pretty sure this is not related. This was passing the tests yesterday. I'll rerun again. |
|
/retest |
|
@Kargakis, thanks for looking into this. |
|
/retest |
|
@kwoodson: The following test failed, say
Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
@sdodson, any recommendations here? This does modify the |
|
The first test failure is just a stale result - The second test failure seems to be flakying a lot in the merge queue (saw it in other PRs too) so I opened the following bugzilla: https://bugzilla.redhat.com/show_bug.cgi?id=1576344 /retest |
This is the forward port of the Azure work that was merged into the 3.9 branch. This work encompasses the following: