Skip to content

[release-4.8] Bug 2012163: Implementing CredentialStoreFactory - #1225

Merged
openshift-merge-robot merged 3 commits into
openshift:release-4.8from
openshift-cherrypick-robot:cherry-pick-1109-to-release-4.8
Oct 8, 2021
Merged

openshift-merge-robot merged 3 commits into
openshift:release-4.8from
openshift-cherrypick-robot:cherry-pick-1109-to-release-4.8

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #1109

/assign dmage

When fetching an image we may need to use different authentications as,
in some cases, we are accessing mirrors instead of the original repo.
Assure we are not reusing the same transport when fetching two distinct
repositories inside the same registry.
@openshift-ci

openshift-ci Bot commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: Bugzilla bug 1972009 has been cloned as Bugzilla bug 2012163. Retitling PR to link against new bug.
/retitle [release-4.8] Bug 2012163: Implementing CredentialStoreFactory

Details

In response to this:

[release-4.8] Bug 1972009: Implementing CredentialStoreFactory

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot changed the title [release-4.8] Bug 1972009: Implementing CredentialStoreFactory [release-4.8] Bug 2012163: Implementing CredentialStoreFactory Oct 8, 2021
@openshift-ci openshift-ci Bot added bugzilla/severity-medium Referenced Bugzilla bug's severity is medium for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. labels Oct 8, 2021
@openshift-ci

openshift-ci Bot commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: This pull request references Bugzilla bug 2012163, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target release (4.8.z) matches configured target release for branch (4.8.z)
  • bug is in the state NEW, which is one of the valid states (NEW, ASSIGNED, ON_DEV, POST, POST)
  • dependent bug Bugzilla bug 1972009 is in the state VERIFIED, which is one of the valid states (VERIFIED, RELEASE_PENDING, CLOSED (ERRATA), CLOSED (CURRENTRELEASE))
  • dependent Bugzilla bug 1972009 targets the "4.9.0" release, which is one of the valid target releases: 4.9.0
  • bug has dependents

Requesting review from QA contact:
/cc @xiuwang

Details

In response to this:

[release-4.8] Bug 2012163: Implementing CredentialStoreFactory

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci
openshift-ci Bot requested a review from xiuwang October 8, 2021 13:23
@dmage

dmage commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci
openshift-ci Bot requested review from smarterclayton and sttts October 8, 2021 13:23
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 8, 2021
@dmage

dmage commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

/assign @sttts

@sttts

sttts commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dmage, openshift-cherrypick-robot, sttts

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 8, 2021
@openshift-merge-robot
openshift-merge-robot merged commit 50376ae into openshift:release-4.8 Oct 8, 2021
@openshift-ci

openshift-ci Bot commented Oct 8, 2021

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: All pull requests linked via external trackers have merged:

Bugzilla bug 2012163 has been moved to the MODIFIED state.

Details

In response to this:

[release-4.8] Bug 2012163: Implementing CredentialStoreFactory

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. bugzilla/severity-medium Referenced Bugzilla bug's severity is medium for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants