Skip to content

Bug 2012163: Supporting mirror authentication during pull through - #297

Merged
openshift-merge-robot merged 2 commits into
openshift:release-4.8from
dmage:credentials-store-factory-4.8
Oct 21, 2021
Merged

openshift-merge-robot merged 2 commits into
openshift:release-4.8from
dmage:credentials-store-factory-4.8

Conversation

@dmage

@dmage dmage commented Oct 18, 2021

Copy link
Copy Markdown
Contributor

Backport of #281.

@openshift-ci openshift-ci Bot added bugzilla/severity-medium Referenced Bugzilla bug's severity is medium for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. labels Oct 18, 2021
@openshift-ci

openshift-ci Bot commented Oct 18, 2021

Copy link
Copy Markdown
Contributor

@dmage: This pull request references Bugzilla bug 2012163, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target release (4.8.z) matches configured target release for branch (4.8.z)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, ON_DEV, POST, POST)
  • dependent bug Bugzilla bug 1972009 is in the state RELEASE_PENDING, which is one of the valid states (VERIFIED, RELEASE_PENDING, CLOSED (ERRATA), CLOSED (CURRENTRELEASE))
  • dependent Bugzilla bug 1972009 targets the "4.9.0" release, which is one of the valid target releases: 4.9.0, 4.9.z
  • bug has dependents

Requesting review from QA contact:
/cc @xiuwang

Details

In response to this:

Bug 2012163: Supporting mirror authentication during pull through

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 18, 2021
@dmage

dmage commented Oct 18, 2021

Copy link
Copy Markdown
Contributor Author

/assign @ricardomaraschini

@ricardomaraschini ricardomaraschini left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 18, 2021
@openshift-ci

openshift-ci Bot commented Oct 18, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dmage, ricardomaraschini

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest-required

Please review the full test history for this PR and help us cut down flakes.

2 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest-required

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest-required

Please review the full test history for this PR and help us cut down flakes.

@xiuwang

xiuwang commented Oct 19, 2021

Copy link
Copy Markdown

Launch cluster from pr.

Steps to validate:
1.Create a restricted cluster
2.Mirror image to mirror registry(enabled auth)
3.Create pullthough imagestream
4.Deploy pod using the pullthough imagestream

/label qe-approved

@openshift-ci openshift-ci Bot added the qe-approved Signifies that QE has signed off on this PR label Oct 19, 2021
@mrunalp

mrunalp commented Oct 20, 2021

Copy link
Copy Markdown
Member

Could you get a backport-risk-assessed from the team?

@dmage

dmage commented Oct 20, 2021

Copy link
Copy Markdown
Contributor Author

/label backport-risk-assessed

@openshift-ci

openshift-ci Bot commented Oct 20, 2021

Copy link
Copy Markdown
Contributor

@dmage: Can not set label backport-risk-assessed: Must be member in one of these teams: [openshift-patch-managers]

Details

In response to this:

/label backport-risk-assessed

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@dmage

dmage commented Oct 20, 2021 •

Copy link
Copy Markdown
Contributor Author

¯\_(ツ)_/¯

@dmage

dmage commented Oct 20, 2021

Copy link
Copy Markdown
Contributor Author

We need to update configs for our repos

@dmage

dmage commented Oct 21, 2021

Copy link
Copy Markdown
Contributor Author

/label backport-risk-assessed

@openshift-ci openshift-ci Bot added the backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. label Oct 21, 2021
@mrunalp mrunalp added the cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. label Oct 21, 2021
@openshift-merge-robot
openshift-merge-robot merged commit 369fd30 into openshift:release-4.8 Oct 21, 2021
@openshift-ci

openshift-ci Bot commented Oct 21, 2021

Copy link
Copy Markdown
Contributor

@dmage: All pull requests linked via external trackers have merged:

Bugzilla bug 2012163 has been moved to the MODIFIED state.

Details

In response to this:

Bug 2012163: Supporting mirror authentication during pull through

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. bugzilla/severity-medium Referenced Bugzilla bug's severity is medium for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. lgtm Indicates that a PR is ready to be merged. qe-approved Signifies that QE has signed off on this PR

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants