Skip to content

channels/fast-4.9: Promote 4.9.10 - #1258

Merged
openshift-merge-robot merged 1 commit into
masterfrom
promote-4.9.10-to-fast-4.9
Dec 7, 2021
Merged

channels/fast-4.9: Promote 4.9.10#1258
openshift-merge-robot merged 1 commit into
masterfrom
promote-4.9.10-to-fast-4.9

Conversation

@vrutkovs

@vrutkovs vrutkovs commented Dec 7, 2021

Copy link
Copy Markdown

It was promoted to the feeder candidate-4.9 by ed92b33 (Merge pull request #1251 from openshift/pr-candidate-4.9.10, 2021-12-02) 5 days, 11:09:39.073332 ago. https://access.redhat.com/errata/RHBA-2021:4889 is public.

It was promoted to the feeder candidate-4.9 by ed92b33 (Merge pull
request #1251 from openshift/pr-candidate-4.9.10, 2021-12-02) 5 days,
11:09:39.073332 ago. https://access.redhat.com/errata/RHBA-2021:4889 is
public.
@vrutkovs vrutkovs added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. labels Dec 7, 2021
@openshift-ci

openshift-ci Bot commented Dec 7, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 0667db8 into master Dec 7, 2021
@wking
wking deleted the promote-4.9.10-to-fast-4.9 branch December 7, 2021 19:47
wking added a commit to wking/cincinnati-graph-data that referenced this pull request Dec 9, 2021
It was promoted to the feeder fast-4.9 by 0667db8 (Merge pull
request openshift#1258 from openshift/promote-4.9.10-to-fast-4.9, 2021-12-07),
but the stabilization bot will have trouble looking beyond the later
move to a version-agnostic fast to figure that out.
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants