Skip to content

Enable 4.9.10 in candidate channel(s) - #1251

Merged
openshift-merge-robot merged 1 commit into
masterfrom
pr-candidate-4.9.10
Dec 2, 2021
Merged

openshift-merge-robot merged 1 commit into
masterfrom
pr-candidate-4.9.10

Conversation

@openshift-bot

Copy link
Copy Markdown
Contributor

This is a release candidate. There is no advisory associated.

Please merge immediately.

@openshift-bot openshift-bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. labels Dec 2, 2021
@openshift-ci

openshift-ci Bot commented Dec 2, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: openshift-bot

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@openshift-ci

openshift-ci Bot commented Dec 2, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: openshift-bot

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit ed92b33 into master Dec 2, 2021
vrutkovs pushed a commit that referenced this pull request Dec 7, 2021
It was promoted to the feeder candidate-4.9 by ed92b33 (Merge pull
request #1251 from openshift/pr-candidate-4.9.10, 2021-12-02) 5 days,
11:09:39.073332 ago. https://access.redhat.com/errata/RHBA-2021:4889 is
public.
@wking
wking deleted the pr-candidate-4.9.10 branch December 7, 2021 19:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants