Skip to content

fix: Properly throw errors on Android now on RN 0.80 - #741

Merged
mrousavy merged 1 commit into
mainfrom
fix/remove-android-error-hacks
Jul 15, 2025
Merged

mrousavy merged 1 commit into
mainfrom
fix/remove-android-error-hacks

Conversation

@mrousavy

@mrousavy mrousavy commented Jul 15, 2025 •

Copy link
Copy Markdown
Member

This requires RN 0.80, but errors in Android are finally fully propagated to JS!

@vercel

vercel Bot commented Jul 15, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
nitro-docs ⬜️ Skipped (Inspect) Jul 15, 2025 3:28pm

This branch was previously deployed

1 inactive deployment
Preview — 3448566e Deployed Jul 15, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Errors on Android don't have a message: Unknown std::runtime_error error

1 participant