Skip to content

[release/11.0.1xx-rc1] Source code updates from microsoft/vstest - #8440

Merged
dotnet-maestro[bot] merged 1 commit into
release/11.0.1xx-rc1from
darc-release/11.0.1xx-rc1-02b3bc9b-16b9-4c8b-9a13-47100c123e56
Aug 19, 2026
Merged

dotnet-maestro[bot] merged 1 commit into
release/11.0.1xx-rc1from
darc-release/11.0.1xx-rc1-02b3bc9b-16b9-4c8b-9a13-47100c123e56

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/microsoft/vstest

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/microsoft/vstest:7e128f4236ebf5f539dc235b367dd954f9c81aeb..https://github.com/dotnet/dotnet:darc-release/11.0.1xx-rc1-02b3bc9b-16b9-4c8b-9a13-47100c123e56

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Important

⚠️ No subscription flowing code in the opposite direction is set up.
Code changes made in the VMR's src/ directory will not flow back to the source repository on branch "main".

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
1 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team August 19, 2026 12:47

@maestro-approver maestro-approver Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This pull request contains only source updates from https://github.com/microsoft/vstest.

@dotnet-maestro
dotnet-maestro Bot merged commit c91260f into release/11.0.1xx-rc1 Aug 19, 2026
16 checks passed
@dotnet-maestro
dotnet-maestro Bot deleted the darc-release/11.0.1xx-rc1-02b3bc9b-16b9-4c8b-9a13-47100c123e56 branch August 19, 2026 15:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants