Reland "WebNN: Support rankRange for op output tensors in opSupportLimits" - #54932
Merged
Merged
Conversation
chromium-wpt-export-bot
marked this pull request as ready for review
September 19, 2025 00:29
…mits" This is a reland of commit 6591b48a057a42cc1a621b2e7b74aa35157c2bd5 This reland fixes WPT failures for CoreML by keeping the error message consistent and adding the WPT rebaseline results. Because blink-side can validate the rank range of operation output, graph input and output, this reland removes the async error handling for reshape output, graph input and output in CoreML backend. Original change's description: > WebNN: Support rankRange for op output tensors in opSupportLimits > > This CL prototypes WebNN spec change [1] that supports tensor rank > range for graph input, constant, output and each operation's output. > > This CL adds new shape rank validation according to allowed output > tensor rank of expand and reshape operations. > > The sequence output of gru and cell operations has different rank, > this CL extends context properties to support them. > > For logical ops, the output rank ranges are set to each op's input > rank ranges, while the output data types are set to unified logical > output data type of context properties. > > [1]: webmachinelearning/webnn#857 > > Bug: 442209350 > Change-Id: Ie5ca5c794cebb4586bcccaaa8d237be1dffea458 > Reviewed-on: https://chromium-review.googlesource.com/c/chromium/src/+/6903617 > Reviewed-by: Reilly Grant <reillyg@chromium.org> > Reviewed-by: Alex Gough <ajgo@chromium.org> > Commit-Queue: ningxin hu <ningxin.hu@intel.com> > Cr-Commit-Position: refs/heads/main@{#1516982} Bug: 442209350, 445793611 Change-Id: Id666c89b8860abe06bac9a1e912057887c3fe03f Cq-Include-Trybots: luci.chromium.try:win11-blink-rel, mac14.arm64-blink-rel, mac14-blink-rel, mac15.arm64-blink-rel, mac15-blink-rel, linux-blink-rel Reviewed-on: https://chromium-review.googlesource.com/c/chromium/src/+/6963704 Reviewed-by: Alex Gough <ajgo@chromium.org> Reviewed-by: Reilly Grant <reillyg@chromium.org> Commit-Queue: ningxin hu <ningxin.hu@intel.com> Cr-Commit-Position: refs/heads/main@{#1517665}
chromium-wpt-export-bot
force-pushed
the
chromium-export-cl-6963704
branch
from
September 19, 2025 00:29
598d131 to
0335a45
Compare
wpt-pr-bot
approved these changes
Sep 19, 2025
wpt-pr-bot
left a comment
Collaborator
There was a problem hiding this comment.
The review process for this patch is being conducted in the Chromium project.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This is a reland of commit 6591b48a057a42cc1a621b2e7b74aa35157c2bd5
This reland fixes WPT failures for CoreML by keeping the error message
consistent and adding the WPT rebaseline results.
Because blink-side can validate the rank range of operation output,
graph input and output, this reland removes the async error handling
for reshape output, graph input and output in CoreML backend.
Original change's description:
Bug: 442209350, 445793611
Change-Id: Id666c89b8860abe06bac9a1e912057887c3fe03f
Cq-Include-Trybots: luci.chromium.try:win11-blink-rel, mac14.arm64-blink-rel, mac14-blink-rel, mac15.arm64-blink-rel, mac15-blink-rel, linux-blink-rel
Reviewed-on: https://chromium-review.googlesource.com/c/chromium/src/+/6963704
Reviewed-by: Alex Gough <ajgo@chromium.org>
Reviewed-by: Reilly Grant <reillyg@chromium.org>
Commit-Queue: ningxin hu <ningxin.hu@intel.com>
Cr-Commit-Position: refs/heads/main@{#1517665}