Skip to content

fix: avoid file import that is incompatible with Windows - #13495

Merged
teemingc merged 4 commits into
sveltejs:mainfrom
Aldthony:netlify-file-import-windows
Feb 24, 2025
Merged

teemingc merged 4 commits into
sveltejs:mainfrom
Aldthony:netlify-file-import-windows

Conversation

@Aldthony

Copy link
Copy Markdown
Contributor

The file import from an absolute path without a file:/// isn't compatible with Windows, and it's also unnecessary since the content is generated a few lines above.


Please don't delete this checklist! Before submitting the PR, please make sure you do the following:

  • It's really useful if your PR references an issue where it is discussed ahead of time. In many cases, features are absent for a reason. For large changes, please create an RFC: https://github.com/sveltejs/rfcs
  • This message body should clearly illustrate what problems it solves.
  • Ideally, include a test that fails without this PR but passes with it.

Tests

  • Run the tests with pnpm test and lint the project with pnpm lint and pnpm check

Changesets

  • If your PR makes a change that should be noted in one or more packages' changelogs, generate a changeset by running pnpm changeset and following the prompts. Changesets that add features should be minor and those that fix bugs should be patch. Please prefix changeset messages with feat:, fix:, or chore:.

Edits

  • Please ensure that 'Allow edits from maintainers' is checked. PRs without this option may be closed.

@changeset-bot

changeset-bot Bot commented Feb 23, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 85389e2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@sveltejs/adapter-netlify Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@svelte-docs-bot

Copy link
Copy Markdown

Comment thread packages/adapter-netlify/index.js Outdated
Comment thread .changeset/eight-walls-melt.md Outdated
@teemingc

Copy link
Copy Markdown
Member

Thanks!

@teemingc
teemingc merged commit 51fc810 into sveltejs:main Feb 24, 2025
@github-actions github-actions Bot mentioned this pull request Feb 24, 2025
Copilot AI pushed a commit to Stadly/kit that referenced this pull request Mar 6, 2026
)

* fix: avoid file import that is incompatible with Windows

* add changeset

* Update packages/adapter-netlify/index.js

* Update .changeset/eight-walls-melt.md

---------

Co-authored-by: Tee Ming <chewteeming01@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants