Skip to content

feat: implements .jsonc support to adapter-cloudflare-workers - #13467

Merged
teemingc merged 15 commits into
sveltejs:mainfrom
duducpp:feat/jsonc-parse
Feb 21, 2025
Merged

teemingc merged 15 commits into
sveltejs:mainfrom
duducpp:feat/jsonc-parse

Conversation

@duducpp

@duducpp duducpp commented Feb 14, 2025 •

Copy link
Copy Markdown
Contributor

resolves #13465

This PR implementes the same jsonc-parser used in cloudflare wrangler


Please don't delete this checklist! Before submitting the PR, please make sure you do the following:

  • It's really useful if your PR references an issue where it is discussed ahead of time. In many cases, features are absent for a reason. For large changes, please create an RFC: https://github.com/sveltejs/rfcs
  • This message body should clearly illustrate what problems it solves.
  • Ideally, include a test that fails without this PR but passes with it.

Tests

  • Run the tests with pnpm test and lint the project with pnpm lint and pnpm check

Changesets

  • If your PR makes a change that should be noted in one or more packages' changelogs, generate a changeset by running pnpm changeset and following the prompts. Changesets that add features should be minor and those that fix bugs should be patch. Please prefix changeset messages with feat:, fix:, or chore:.

Edits

  • Please ensure that 'Allow edits from maintainers' is checked. PRs without this option may be closed.

@changeset-bot

changeset-bot Bot commented Feb 14, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d758e05

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@sveltejs/adapter-cloudflare-workers Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@svelte-docs-bot

Copy link
Copy Markdown

@duducpp duducpp changed the title Feat/jsonc parse feat: jsonc parse Feb 14, 2025
@duducpp duducpp changed the title feat: jsonc parse feat: jsonc-parse Feb 14, 2025
@duducpp duducpp closed this Feb 14, 2025
@duducpp duducpp reopened this Feb 14, 2025
@duducpp duducpp changed the title feat: jsonc-parse feat: implements .jsonc support to adapter-cloudflare-workers Feb 14, 2025
@benmccann

Copy link
Copy Markdown
Member

This would need to be in adapter-cloudflare as well as we normally try to keep the two in sync

removes jsonc-parser dependency
@duducpp

duducpp commented Feb 14, 2025

Copy link
Copy Markdown
Contributor Author

This would need to be in adapter-cloudflare as well as we normally try to keep the two in sync

I also tried updating it in adapter-cloudflare but I couldnt find any references to the wrangler config file on it.

@teemingc teemingc left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thank you! I've updated the docs as well (and fixed some links). It seems like Cloudflare is also pushing for wrangler.jsonc files over wrangler.toml files in their docs

@teemingc
teemingc merged commit 908a948 into sveltejs:main Feb 21, 2025
@github-actions github-actions Bot mentioned this pull request Feb 21, 2025
@duducpp
duducpp deleted the feat/jsonc-parse branch February 25, 2025 15:12
Copilot AI pushed a commit to Stadly/kit that referenced this pull request Mar 6, 2026
…js#13467)

* feat: implements jsonc-parse

* feat: updates adapter-cloudflare-workers documentation

* feat: using vite approach to get wrangler config

removes jsonc-parser dependency

* update the docs

* remove toml dep

* add adapter option description for config

* allow wrangler to find the config file

* add changeset

* json and jsonc requires wrangler 3.91.0

* standardise troubleshooting sections

* fix lock file

* update links

* fix pnpm lock

---------

Co-authored-by: Chew Tee Ming <chew.tee.ming@nindatech.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

adapter-cloudflare-workers not parsing default .jsonc

3 participants