Skip to content

bump vite to 8.3 - #1291

Merged
jycouet merged 4 commits into
sveltejs:version-1from
sacrosanctic:vite-8.3
Sep 19, 2026
Merged

jycouet merged 4 commits into
sveltejs:version-1from
sacrosanctic:vite-8.3

Conversation

@sacrosanctic

Copy link
Copy Markdown
Contributor

Closes #1288

Description

pending https://github.com/vitejs/vite/milestone/34

Checklist

  • Update snapshots (if applicable)
  • Add a changeset (if applicable)
  • Allow maintainers to edit this PR
  • I care about what I'm doing, no matter the tool I use (Notepad, Sublime, VSCode, AI...)

@pkg-svelte-dev

pkg-svelte-dev Bot commented Aug 21, 2026

Copy link
Copy Markdown

Install the latest version of sv from 82e4aad:

pnx https://pkg.svelte.dev/sv/c/82e4aad7e75cd6db95c273df5367e90ec8af87d6 create

Open in pkg.svelte.dev: https://pkg.svelte.dev/repos/cli/pr/1291

Note

This PR is from a fork. A maintainer must approve approve each commit before it can be built and installed.

@changeset-bot

changeset-bot Bot commented Aug 21, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 82e4aad

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
sv Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@sacrosanctic sacrosanctic changed the title bump to 8.3 bump vite to 8.3 Aug 22, 2026
@sacrosanctic
sacrosanctic marked this pull request as ready for review September 19, 2026 13:01

@jycouet jycouet left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thank you

@jycouet
jycouet merged commit a6c3b0c into sveltejs:version-1 Sep 19, 2026
7 of 8 checks passed
@sacrosanctic
sacrosanctic deleted the vite-8.3 branch September 19, 2026 14:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

SvelteKit 3 migrate: use relative paths for dynamic imports

2 participants