Automate config brancher by auto-config-brancher job at Tue, 18 Aug 2026 08:11:02 UTC - #83539
Conversation
d949643 to
da0f7c9
Compare
|
New changes are detected. LGTM label has been removed. |
da0f7c9 to
ed043ff
Compare
|
New changes are detected. LGTM label has been removed. |
ed043ff to
1a96c78
Compare
|
New changes are detected. LGTM label has been removed. |
….1 --skip-periodics --future-release 4.23 --future-release 5.2 --confirm
…openshift-priv --only-org openshift --whitelist-file ./core-services/openshift-priv/_whitelist.yaml
…rator/jobs --registry ./ci-operator/step-registry
…/core-services/sanitize-prow-jobs/_config.yaml
1a96c78 to
dd6b91a
Compare
|
New changes are detected. LGTM label has been removed. |
|
[APPROVALNOTIFIER] This PR is APPROVED Approval requirements bypassed by manually added approval. This pull-request has been approved by: openshift-merge-bot[bot] The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
[REHEARSALNOTIFIER]
A total of 40 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs. A full list of affected jobs can be found here Interacting with pj-rehearseComment: Once you are satisfied with the results of the rehearsals, comment: |
|
@openshift-merge-bot[bot]: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
/cc @openshift/test-platform