Skip to content

Automate config brancher by auto-config-brancher job at Mon, 17 Aug 2026 08:09:47 UTC - #82927

Merged
openshift-merge-robot merged 6 commits into
mainfrom
auto-config-brancher
Aug 17, 2026
Merged

Automate config brancher by auto-config-brancher job at Mon, 17 Aug 2026 08:09:47 UTC#82927
openshift-merge-robot merged 6 commits into
mainfrom
auto-config-brancher

Conversation

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/cc @openshift/test-platform

@openshift-merge-bot openshift-merge-bot Bot added tide/merge-method-merge Denotes a PR that should use a standard merge by tide when it merges. rehearsals-ack Signifies that rehearsal jobs have been acknowledged priority/ci-critical Critical priority - CI infrastructure is broken or severely impacted approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Aug 4, 2026
@openshift-ci
openshift-ci Bot requested a review from a team August 4, 2026 20:12
@openshift-merge-bot openshift-merge-bot Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 4, 2026
@openshift-merge-bot openshift-merge-bot Bot changed the title Automate config brancher by auto-config-brancher job at Tue, 04 Aug 2026 20:11:54 UTC Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 00:10:50 UTC Aug 5, 2026
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Aug 5, 2026
@openshift-ci

openshift-ci Bot commented Aug 5, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-merge-bot openshift-merge-bot Bot removed the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Aug 5, 2026
@openshift-merge-bot openshift-merge-bot Bot changed the title Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 00:10:50 UTC Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 04:08:46 UTC Aug 5, 2026
@openshift-merge-bot openshift-merge-bot Bot added rehearsals-ack Signifies that rehearsal jobs have been acknowledged lgtm Indicates that a PR is ready to be merged. labels Aug 5, 2026
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Aug 5, 2026
@openshift-ci

openshift-ci Bot commented Aug 5, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-merge-bot openshift-merge-bot Bot removed the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Aug 5, 2026
@openshift-merge-bot openshift-merge-bot Bot changed the title Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 04:08:46 UTC Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 08:11:21 UTC Aug 5, 2026
@openshift-merge-bot openshift-merge-bot Bot added rehearsals-ack Signifies that rehearsal jobs have been acknowledged lgtm Indicates that a PR is ready to be merged. labels Aug 5, 2026
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Aug 5, 2026
@openshift-ci

openshift-ci Bot commented Aug 5, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-merge-bot openshift-merge-bot Bot removed the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Aug 5, 2026
@openshift-merge-bot openshift-merge-bot Bot changed the title Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 08:11:21 UTC Automate config brancher by auto-config-brancher job at Wed, 05 Aug 2026 12:14:22 UTC Aug 5, 2026
@openshift-merge-bot openshift-merge-bot Bot added rehearsals-ack Signifies that rehearsal jobs have been acknowledged lgtm Indicates that a PR is ready to be merged. labels Aug 5, 2026
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Aug 5, 2026
@openshift-ci

openshift-ci Bot commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

18 similar comments
@openshift-ci

openshift-ci Bot commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 13, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 13, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 13, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 13, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 13, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 13, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 14, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 14, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 14, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 14, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 14, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 14, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

….1 --skip-periodics --future-release 4.23 --future-release 5.2 --confirm
…openshift-priv --only-org openshift --whitelist-file ./core-services/openshift-priv/_whitelist.yaml
…rator/jobs --registry ./ci-operator/step-registry
…865 --github-app-private-key-path /etc/github/cert --github-endpoint http://ghproxy --dry-run=false --whitelist-file ./core-services/openshift-priv/_whitelist.yaml
…/core-services/sanitize-prow-jobs/_config.yaml
@openshift-ci

openshift-ci Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: openshift-merge-bot[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot

Copy link
Copy Markdown
Contributor Author

[REHEARSALNOTIFIER]
@openshift-merge-bot[bot]: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-cluster-api-release-4.23-e2e-aws-capi-techpreview openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-aws-ovn openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-aws-ovn-techpreview openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-aws-serial-1of2 openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-aws-serial-2of2 openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-azure-capi-techpreview openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-gcp-capi-techpreview openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-hypershift openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-openstack-capi-techpreview openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-vsphere-capi-techpreview openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-e2e-vsphere-staticip-ipam-ovn openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-images openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-regression-clusterinfra-aws-ipi-techpreview-capi openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-unit openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-verify openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-verify-commits openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-cluster-api-release-4.23-verify-deps openshift/cluster-api presubmit Ci-operator config changed
pull-ci-openshift-api-release-4.23-verify-hypershift-integration openshift/api presubmit Ci-operator config changed
pull-ci-openshift-jobset-operator-release-4.23-e2e-aws-operator-serial-ote openshift/jobset-operator presubmit Presubmit changed
pull-ci-openshift-aws-encryption-provider-release-4.23-e2e-hypershift openshift/aws-encryption-provider presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-release-4.23-artifacts openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-release-4.23-configmap-scale openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-release-4.23-e2e-agnostic-ovn-cmd openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-release-4.23-e2e-aws openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-release-4.23-e2e-aws-csi openshift/kubernetes presubmit Ci-operator config changed

A total of 666 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@openshift-ci

openshift-ci Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

@openshift-merge-bot[bot]: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/check-gh-automation 8dee913 link true /test check-gh-automation
ci/prow/check-gh-automation-tide 8dee913 link true /test check-gh-automation-tide
ci/prow/config 8dee913 link true /test config

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. priority/ci-critical Critical priority - CI infrastructure is broken or severely impacted tide/merge-method-merge Denotes a PR that should use a standard merge by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants