Skip to content

PowerVS: Add more debugging information to verbose DHCP output - #66412

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
hamzy:PowerVS-add-no-dhcp-msg
Jun 27, 2025
Merged

PowerVS: Add more debugging information to verbose DHCP output#66412
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
hamzy:PowerVS-add-no-dhcp-msg

Conversation

@hamzy

@hamzy hamzy commented Jun 24, 2025

Copy link
Copy Markdown
Contributor

Add more debugging information to verbose DHCP output.

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 24, 2025
@openshift-ci
openshift-ci Bot requested review from Basavaraju-G and mkumatag June 24, 2025 22:14
@hamzy

hamzy commented Jun 24, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from 5a6cfd9 to 4bbec36 Compare June 25, 2025 18:40
@hamzy

hamzy commented Jun 25, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy

hamzy commented Jun 25, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy

hamzy commented Jun 25, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from 4bbec36 to 1597431 Compare June 25, 2025 22:40
@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from 1597431 to 037fd53 Compare June 26, 2025 10:28
@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from 037fd53 to 838296b Compare June 26, 2025 12:08
@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from 838296b to bf15d8c Compare June 26, 2025 13:21
@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from bf15d8c to 71f4c58 Compare June 26, 2025 14:39
@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

Add more debugging information to verbose DHCP output.
@hamzy
hamzy force-pushed the PowerVS-add-no-dhcp-msg branch from 71f4c58 to 09ea370 Compare June 26, 2025 18:26
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@hamzy: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-installer-main-e2e-powervs-capi-ovn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.21-e2e-powervs-capi-ovn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.20-e2e-powervs-capi-ovn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.19-e2e-powervs-capi-ovn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.18-e2e-powervs-capi-ovn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.17-e2e-powervs-ovn openshift/installer presubmit Registry content changed
pull-ci-openshift-installer-release-4.16-altinfra-e2e-powervs-capi-ovn openshift/installer presubmit Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.16-ppc64le-nightly-powervs-ipi-f28-destructive N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.17-ppc64le-nightly-powervs-ipi-f14-destructive N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.20-ppc64le-nightly-powervs-ipi-f7-destructive N/A periodic Registry content changed
periodic-ci-openshift-release-master-ci-4.16-e2e-powervs-ovn-techpreview N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.14-ocp-e2e-ovn-ppc64le-powervs N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-ppc64le-nightly-powervs-ipi-f7 N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.14-ocp-e2e-serial-ovn-ppc64le-powervs N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.18-ppc64le-nightly-powervs-ipi-f7 N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.17-ocp-e2e-ovn-powervs-capi-multi-p-p N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.19-ocp-e2e-ovn-powervs-capi-multi-p-p N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.19-ppc64le-nightly-powervs-ipi-f7-destructive N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.18-ppc64le-nightly-powervs-ipi-f7-destructive N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.17-ppc64le-nightly-powervs-ipi-f14 N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.16-ocp-e2e-ovn-ppc64le-powervs-original N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.13-ocp-e2e-serial-ovn-ppc64le-powervs N/A periodic Registry content changed
periodic-ci-openshift-openshift-tests-private-release-4.20-ppc64le-nightly-powervs-ipi-f7 N/A periodic Registry content changed
periodic-ci-openshift-multiarch-master-nightly-4.15-ocp-e2e-ovn-ppc64le-powervs-original N/A periodic Registry content changed

A total of 29 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse periodic-ci-openshift-multiarch-master-nightly-4.20-ocp-e2e-ovn-powervs-capi-multi-p-p

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@hamzy

hamzy commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@hamzy: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Jun 26, 2025
@tvardema

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jun 27, 2025
@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hamzy, tvardema

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

@hamzy: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 788639d into openshift:master Jun 27, 2025
liweinan pushed a commit to liweinan/release that referenced this pull request Aug 5, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants