Skip to content

Onboarding mtv-integrations to OpenShift CI - #66134

Merged
openshift-merge-bot[bot] merged 9 commits into
openshift:masterfrom
sahare:add-mtv-integrations
Jun 27, 2025
Merged

Onboarding mtv-integrations to OpenShift CI#66134
openshift-merge-bot[bot] merged 9 commits into
openshift:masterfrom
sahare:add-mtv-integrations

Conversation

@sahare

@sahare sahare commented Jun 16, 2025

Copy link
Copy Markdown
Contributor

sahare added 3 commits June 16, 2025 10:58
Signed-off-by: sahare <sahare@redhat.com>
Signed-off-by: sahare <sahare@redhat.com>
Signed-off-by: sahare <sahare@redhat.com>
@openshift-ci
openshift-ci Bot requested review from gurnben and smithbw88 June 16, 2025 18:12
@sahare

sahare commented Jun 17, 2025

Copy link
Copy Markdown
Contributor Author

/test ordered-prow-config

@sahare

sahare commented Jun 17, 2025

Copy link
Copy Markdown
Contributor Author

/cc @jnpacker

@openshift-ci
openshift-ci Bot requested a review from jnpacker June 17, 2025 18:04
@sahare

sahare commented Jun 17, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 17, 2025

Copy link
Copy Markdown
Contributor Author

/retest

@sahare

sahare commented Jun 17, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-images

@sahare

sahare commented Jun 17, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

1 similar comment
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/retest

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 18, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

Signed-off-by: sahare <sahare@redhat.com>
@sahare

sahare commented Jun 19, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@sahare

sahare commented Jun 19, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

Signed-off-by: sahare <sahare@redhat.com>
@sahare

sahare commented Jun 19, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 19, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@sahare

sahare commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@smithbw88 smithbw88 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@openshift-ci openshift-ci Bot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Jun 26, 2025
Signed-off-by: sahare <sahare@redhat.com>
@sahare

sahare commented Jun 26, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Jun 26, 2025
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@sahare: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-stolostron-mtv-integrations-main-images stolostron/mtv-integrations presubmit Presubmit changed
pull-ci-stolostron-mtv-integrations-main-sonar stolostron/mtv-integrations presubmit Presubmit changed
pull-ci-stolostron-mtv-integrations-main-unit-tests stolostron/mtv-integrations presubmit Presubmit changed
Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@sahare

sahare commented Jun 27, 2025

Copy link
Copy Markdown
Contributor Author

The job is now failing solely due to a failed SonarCloud Quality Gate, not due to misconfiguration or permission errors. SonarCloud successfully analyzed the code but determined that the results do not meet the project's defined quality standards.

@sahare

sahare commented Jun 27, 2025

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sahare: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Jun 27, 2025
@jnpacker

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jun 27, 2025
@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jnpacker, sahare, smithbw88

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 6794b25 into openshift:master Jun 27, 2025
@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

@sahare: Updated the config configmap in namespace ci at cluster app.ci using the following files:

  • key core-services-prow-02_config-stolostron-mtv-integrations-_prowconfig.yaml using file core-services/prow/02_config/stolostron/mtv-integrations/_prowconfig.yaml
Details

In response to this:

https://issues.redhat.com/browse/ACM-21067

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci

openshift-ci Bot commented Jun 27, 2025

Copy link
Copy Markdown
Contributor

@sahare: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/stolostron/mtv-integrations/main/sonar 3cca860 link unknown /pj-rehearse pull-ci-stolostron-mtv-integrations-main-sonar

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

liweinan pushed a commit to liweinan/release that referenced this pull request Aug 5, 2025
* initial prow onboarding

Signed-off-by: sahare <sahare@redhat.com>

* make update successful

Signed-off-by: sahare <sahare@redhat.com>

* add sonar post submit

Signed-off-by: sahare <sahare@redhat.com>

* customize SONAR_GO_TEST_ARGS to exclude test/e2e

Signed-off-by: sahare <sahare@redhat.com>

* script syntax error during command evaluation

Signed-off-by: sahare <sahare@redhat.com>

* modify commands blocks for both sonar and sonar-post-submit

Signed-off-by: sahare <sahare@redhat.com>

* temp change to debug

Signed-off-by: sahare <sahare@redhat.com>

* trying to fix the command to exclude e2e

Signed-off-by: sahare <sahare@redhat.com>

* setting the user.home system property for the JVM

Signed-off-by: sahare <sahare@redhat.com>

---------

Signed-off-by: sahare <sahare@redhat.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants