Skip to content

ci-operator/templates/openshift: Drop 'openshift-install version' - #2962

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:drop-openshift-install-version
Feb 27, 2019
Merged

ci-operator/templates/openshift: Drop 'openshift-install version'#2962
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:drop-openshift-install-version

Conversation

@wking

@wking wking commented Feb 26, 2019

Copy link
Copy Markdown
Member

I'd initially added this in eba17a1 (#1981), but it's redundant since openshift/installer@3d8019f9 (openshift/installer#1256).

I'd initially added this in eba17a1 (cluster-launch-installer-e2e:
Log the installer version, 2018-10-19, openshift#1981), but it's redundant
since openshift/installer@3d8019f9 (openshift-install: add version
string to logs, 2019-02-14, openshift/installer#1256).
@openshift-ci-robot openshift-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Feb 26, 2019
@wking

wking commented Feb 26, 2019

Copy link
Copy Markdown
Member Author

CC @vrutkovs, @sdodson

@vrutkovs

Copy link
Copy Markdown
Contributor

/approve
for openshift-ansible part

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 27, 2019
@sdodson

sdodson commented Feb 27, 2019

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Feb 27, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sdodson, vrutkovs, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 17a8e84 into openshift:master Feb 27, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wking: Updated the following 8 configmaps:

  • prow-job-cluster-launch-installer-e2e configmap in namespace ci-stg using the following files:
    • key cluster-launch-installer-e2e.yaml using file ci-operator/templates/openshift/installer/cluster-launch-installer-e2e.yaml
  • prow-job-cluster-launch-installer-src configmap in namespace ci using the following files:
    • key cluster-launch-installer-src.yaml using file ci-operator/templates/openshift/installer/cluster-launch-installer-src.yaml
  • prow-job-cluster-launch-installer-src configmap in namespace ci-stg using the following files:
    • key cluster-launch-installer-src.yaml using file ci-operator/templates/openshift/installer/cluster-launch-installer-src.yaml
  • prow-job-cluster-launch-e2e-40 configmap in namespace ci using the following files:
    • key cluster-launch-e2e-40.yaml using file ci-operator/templates/openshift/openshift-ansible/cluster-launch-e2e-40.yaml
  • prow-job-cluster-launch-e2e-40 configmap in namespace ci-stg using the following files:
    • key cluster-launch-e2e-40.yaml using file ci-operator/templates/openshift/openshift-ansible/cluster-launch-e2e-40.yaml
  • prow-job-cluster-scaleup-e2e-40 configmap in namespace ci using the following files:
    • key cluster-scaleup-e2e-40.yaml using file ci-operator/templates/openshift/openshift-ansible/cluster-scaleup-e2e-40.yaml
  • prow-job-cluster-scaleup-e2e-40 configmap in namespace ci-stg using the following files:
    • key cluster-scaleup-e2e-40.yaml using file ci-operator/templates/openshift/openshift-ansible/cluster-scaleup-e2e-40.yaml
  • prow-job-cluster-launch-installer-e2e configmap in namespace ci using the following files:
    • key cluster-launch-installer-e2e.yaml using file ci-operator/templates/openshift/installer/cluster-launch-installer-e2e.yaml
Details

In response to this:

I'd initially added this in eba17a1 (#1981), but it's redundant since openshift/installer@3d8019f9 (openshift/installer#1256).

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@wking
wking deleted the drop-openshift-install-version branch February 27, 2019 14:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants