Skip to content

Update base os image to centos 8 and go version to 1.14 - #11381

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
marun:fix-libgo-go-version
Aug 28, 2020
Merged

Update base os image to centos 8 and go version to 1.14#11381
openshift-merge-robot merged 1 commit into
openshift:masterfrom
marun:fix-libgo-go-version

Conversation

@marun

@marun marun commented Aug 28, 2020

Copy link
Copy Markdown
Contributor

Apparently this is required to get library-go building with golang 1.14, which is required by the 1.19 bump for library-go: openshift/library-go#872

@sttts

sttts commented Aug 28, 2020

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@sttts

sttts commented Aug 28, 2020

Copy link
Copy Markdown
Contributor

/retest

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: marun, sttts

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Aug 28, 2020
@marun

marun commented Aug 28, 2020

Copy link
Copy Markdown
Contributor Author

/test step-registry-metadata

@petr-muller

Copy link
Copy Markdown
Member

/refresh

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@marun: The following test failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/prow/pj-rehearse 57444ee link /test pj-rehearse

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot
openshift-merge-robot merged commit 7db87fb into openshift:master Aug 28, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@marun: Updated the ci-operator-master-configs configmap in namespace ci at cluster app.ci using the following files:

  • key openshift-library-go-master.yaml using file ci-operator/config/openshift/library-go/openshift-library-go-master.yaml
Details

In response to this:

Apparently this is required to get library-go building with golang 1.14, which is required by the 1.19 bump for library-go: openshift/library-go#872

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@marun
marun deleted the fix-libgo-go-version branch February 1, 2021 19:18
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants