Skip to content

NO-JIRA: [bot] Bump openshift/prom-label-proxy to v0.12.1 - #385

Merged
openshift-merge-bot[bot] merged 11 commits into
openshift:mainfrom
rhobs:automated-updates-main
Sep 12, 2025
Merged

openshift-merge-bot[bot] merged 11 commits into
openshift:mainfrom
rhobs:automated-updates-main

Conversation

@openshift-monitoring-bot

Copy link
Copy Markdown

Description

This is an automated version bump from CI.
The logs for this run can be found in the syncbot repo actions.
If you wish to perform this manually, execute the following commands from openshift/prom-label-proxy repo:

git fetch https://github.com/prometheus-community/prom-label-proxy --tags
if ! git merge refs/tags/v0.12.1 --no-edit; then
  git checkout --theirs CHANGELOG.md VERSION go.mod go.sum examples README.md main.go
  git checkout --ours OWNERS
  git add CHANGELOG.md VERSION go.mod go.sum examples README.md main.go OWNERS
  git merge --continue
fi
go mod tidy
go mod vendor

if [ -f scripts/rh-manifest.sh ]; then
  bash scripts/rh-manifest.sh
  git add rh-manifest.txt
  git diff --cached --exit-code || git commit -s -m "[bot] update rh-manifest.txt"
fi

simonpasquier and others added 11 commits August 25, 2025 17:07
Signed-off-by: Simon Pasquier <spasquie@redhat.com>
Bumps [github.com/prometheus/client_golang](https://github.com/prometheus/client_golang) from 1.23.0 to 1.23.2.
- [Release notes](https://github.com/prometheus/client_golang/releases)
- [Changelog](https://github.com/prometheus/client_golang/blob/main/CHANGELOG.md)
- [Commits](prometheus/client_golang@v1.23.0...v1.23.2)

---
updated-dependencies:
- dependency-name: github.com/prometheus/client_golang
  dependency-version: 1.23.2
  dependency-type: direct:production
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
Signed-off-by: prombot <prometheus-team@googlegroups.com>
…/go_modules/github.com/prometheus/client_golang-1.23.2

build(deps): bump github.com/prometheus/client_golang from 1.23.0 to 1.23.2
Signed-off-by: Simon Pasquier <spasquie@redhat.com>
fix: don't panic on error-on-replace with multiple values
Signed-off-by: Simon Pasquier <spasquie@redhat.com>
v0.12.1

# -----BEGIN PGP SIGNATURE-----
#
# iQIzBAABCgAdFiEE/jCAc0MNZY5toBt1AZCmbAoQ/E8FAmjD3L4ACgkQAZCmbAoQ
# /E9FFBAAsgiMQLCeSLuW4yQ1d3X+9+6pSxW5Nn5r+Jn8DxTKnG9ZsfVgoET7xqOw
# K+PO9YwrJrWvTDmfBOSjckq586FpkcYjfBFrNxzfuqrlyMWH+V0nhZ890HWE/QHz
# cG8Wl3zJpF2c6up0M/ywQ8FSwa51EwNCef5UZEZVsThUBjN0tFNrGQfYDiKyI1LK
# YpfJ8izPUEkt894ZqvgWdywu6b6+NuVcQhig4m2dpPB9Hmm4eqbiJZyPr3Qhr+Me
# YLDlUVwBUGQ56hNBzdYiBY92cW1s94YIUn3N9rRevwf1yiSSH26oQSb4MTxuYHcU
# NR8BqF1JPBqaKoexndWMqwWT0QCWyHtbH6tW63V9Tlu0K+XqL4l9zaI6PIRBbte4
# BvbJnJPmiT+I9IgUP4Oulrjm1oiFxvCA3yileOv48xB9h4ZxSE78310Xbs+i8Ful
# r8XsExfKD6tlwTpxBB7ypMccbFVjLTAGCUQMrHqnDs3RW+kusulmCVLT7xlmWkH4
# RR1gPvDCxfY/JJk6ZnjLAwN7sGyswsbrUEL7u/bB7iG8Yzr0YVhQFOgDTvTpd6zb
# mhdklX65MXz0/Lh+UojL1GzphmacluXBKTidYxX3KmKm4lAxYGNWhRoPObHGCaza
# wu3rahT9HqCD3JgYSWsP8Wfa9F6Ax5sJVXQ88jL/F670P/JtSD4=
# =TV3R
# -----END PGP SIGNATURE-----
# gpg: directory '/home/runner/.gnupg' created
# gpg: keybox '/home/runner/.gnupg/pubring.kbx' created
# gpg: Signature made Fri Sep 12 08:41:34 2025 UTC
# gpg:                using RSA key FE308073430D658E6DA01B750190A66C0A10FC4F
# gpg: Can't check signature: No public key
Signed-off-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
@simonpasquier

Copy link
Copy Markdown

/lgtm
/approve
/retitle NO-JIRA: [bot] Bump openshift/prom-label-proxy to v0.12.1

@openshift-ci openshift-ci Bot changed the title [bot] Bump openshift/prom-label-proxy to v0.12.1 NO-JIRA: [bot] Bump openshift/prom-label-proxy to v0.12.1 Sep 12, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@openshift-monitoring-bot[bot]: This pull request explicitly references no jira issue.

Details

In response to this:

Description

This is an automated version bump from CI.
The logs for this run can be found in the syncbot repo actions.
If you wish to perform this manually, execute the following commands from openshift/prom-label-proxy repo:

git fetch https://github.com/prometheus-community/prom-label-proxy --tags
if ! git merge refs/tags/v0.12.1 --no-edit; then
 git checkout --theirs CHANGELOG.md VERSION go.mod go.sum examples README.md main.go
 git checkout --ours OWNERS
 git add CHANGELOG.md VERSION go.mod go.sum examples README.md main.go OWNERS
 git merge --continue
fi
go mod tidy
go mod vendor

if [ -f scripts/rh-manifest.sh ]; then
 bash scripts/rh-manifest.sh
 git add rh-manifest.txt
 git diff --cached --exit-code || git commit -s -m "[bot] update rh-manifest.txt"
fi

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 12, 2025
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 12, 2025
@openshift-ci

openshift-ci Bot commented Sep 12, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: openshift-monitoring-bot[bot], simonpasquier

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 12, 2025
@simonpasquier

Copy link
Copy Markdown

/verified by e2e-agnostic-cmo

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Sep 12, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@simonpasquier: This PR has been marked as verified by e2e-agnostic-cmo.

Details

In response to this:

/verified by e2e-agnostic-cmo

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Sep 12, 2025

Copy link
Copy Markdown

@openshift-monitoring-bot[bot]: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 70f3c88 into openshift:main Sep 12, 2025
9 checks passed
monitoring-commit-bot Bot pushed a commit to rhobs/prom-label-proxy that referenced this pull request Sep 16, 2026
…/go_modules/github.com/prometheus/prometheus-0.313.0

build(deps): bump github.com/prometheus/prometheus from 0.312.0 to 0.313.0
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants