tests: Remove duplicated toolbox test - #1034
Conversation
The new toolbox test is already available through the shared symlink. It was however marked FCOS only and was thus never run. It will thus be picked up here once we update it in f-c-c. Fixes: ebaeae8 tests: Use new toolbox test
|
Note that the test is not working on RHCOS right now. Filed #1035 |
|
All the failures are unrelated. |
|
/test rhcos-86-build-test-qemu |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ravanelli, travier The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/retest |
|
|
/retest |
|
@travier: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
The new toolbox test is already available through the shared symlink. It was however marked FCOS only and was thus never run. It will thus be picked up here once we update it in f-c-c.
Fixes: ebaeae8 tests: Use new toolbox test