Skip to content

Add new tests for baremetal IPI platform - #25853

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
andfasano:baremetal-tests
Mar 21, 2021
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
andfasano:baremetal-tests

Conversation

@andfasano

@andfasano andfasano commented Feb 3, 2021

Copy link
Copy Markdown
Contributor

This PR adds some new basic tests for the baremetal IPI platform, so that it could be possible to integrate them in the current CI/release builds for the baremetal IPI platform.

Tests are referring the following enhancements:

CI baremetal workflow:

This is just an initial PR to begin landing tests for the baremetal IPI platform and clear out the required technical details.

@openshift-ci-robot openshift-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Feb 3, 2021
@andfasano

Copy link
Copy Markdown
Contributor Author

/assign @deads2k

@andfasano andfasano changed the title [WIP] Add new tests for baremetal platform [WIP] Add new tests for baremetal IPI platform Feb 3, 2021
@andfasano

Copy link
Copy Markdown
Contributor Author

/retest

@andfasano andfasano changed the title [WIP] Add new tests for baremetal IPI platform Add new tests for baremetal IPI platform Feb 12, 2021
@openshift-ci-robot openshift-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Feb 12, 2021
@stbenjam

Copy link
Copy Markdown
Member

Will enabling e2e-metal-ipi on this repo let us see the tests pass?

@andfasano andfasano closed this Feb 12, 2021
@andfasano andfasano reopened this Feb 12, 2021
@andfasano

Copy link
Copy Markdown
Contributor Author

Will enabling e2e-metal-ipi on this repo let us see the tests pass?

Good idea, openshift/release#15816

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-metal-ipi

@stbenjam

Copy link
Copy Markdown
Member

Can we get an OWNERS file added to the baremetal directories?

Initial tests look fine, but plan is to add more? I'd like us to make sure control plane hosts are adopted correctly and reflect "Externally provisioned." And if there's no existing test for this here, make sure that all our machines have a node ref.

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-metal-ipi

1 similar comment
@andfasano

andfasano commented Mar 1, 2021

Copy link
Copy Markdown
Contributor Author

/test e2e-metal-ipi

@openshift-ci-robot

Copy link
Copy Markdown

@andfasano: The specified target(s) for /test were not found.
The following commands are available to trigger jobs:

  • /test e2e-agnostic-cmd
  • /test e2e-aws
  • /test e2e-aws-csi
  • /test e2e-aws-disruptive
  • /test e2e-aws-fips
  • /test e2e-aws-image-registry
  • /test e2e-aws-jenkins
  • /test e2e-aws-multitenant
  • /test e2e-aws-ovn
  • /test e2e-aws-proxy
  • /test e2e-aws-serial
  • /test e2e-aws-upgrade
  • /test e2e-azure
  • /test e2e-gcp
  • /test e2e-gcp-builds
  • /test e2e-gcp-csi
  • /test e2e-gcp-disruptive
  • /test e2e-gcp-image-ecosystem
  • /test e2e-gcp-upgrade
  • /test e2e-metal-ipi
  • /test e2e-metal-ipi-ovn-dualstack
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-metal-ipi-virtualmedia
  • /test e2e-openstack
  • /test e2e-openstack-serial
  • /test e2e-vsphere
  • /test images
  • /test okd-e2e-gcp
  • /test verify
  • /test verify-deps
  • /test extended_gssapi
  • /test extended_ldap_groups
  • /test extended_networking

Use /test all to run the following jobs:

  • pull-ci-openshift-origin-master-e2e-agnostic-cmd
  • pull-ci-openshift-origin-master-e2e-aws-csi
  • pull-ci-openshift-origin-master-e2e-aws-disruptive
  • pull-ci-openshift-origin-master-e2e-aws-fips
  • pull-ci-openshift-origin-master-e2e-aws-serial
  • pull-ci-openshift-origin-master-e2e-gcp
  • pull-ci-openshift-origin-master-e2e-gcp-builds
  • pull-ci-openshift-origin-master-e2e-gcp-csi
  • pull-ci-openshift-origin-master-e2e-gcp-disruptive
  • pull-ci-openshift-origin-master-e2e-gcp-upgrade
  • pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-origin-master-images
  • pull-ci-openshift-origin-master-verify
  • pull-ci-openshift-origin-master-verify-deps
Details

In response to this:

/test ci/prow/e2e-metal-ipi

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@andfasano

Copy link
Copy Markdown
Contributor Author

Looks like the new two tests (tagged with [sig-installer][Feature:baremetal]) are correctly executed and passing during the e2e-metal-ipi job

@stbenjam

stbenjam commented Mar 1, 2021

Copy link
Copy Markdown
Member

Looks like the new two tests (tagged with [sig-installer][Feature:baremetal]) are correctly executed and passing during the e2e-metal-ipi job

Nice! Did you see my comment above, specifically the OWNERS file so we can add tests later?

#25853 (comment)

@andfasano

Copy link
Copy Markdown
Contributor Author

Looks like the new two tests (tagged with [sig-installer][Feature:baremetal]) are correctly executed and passing during the e2e-metal-ipi job

Nice! Did you see my comment above, specifically the OWNERS file so we can add tests later?

#25853 (comment)

Thanks, added

@andfasano

andfasano commented Mar 2, 2021

Copy link
Copy Markdown
Contributor Author

/test all

@openshift-ci-robot

Copy link
Copy Markdown

@andfasano: The specified target(s) for /test were not found.
The following commands are available to trigger jobs:

  • /test e2e-agnostic-cmd
  • /test e2e-aws
  • /test e2e-aws-csi
  • /test e2e-aws-disruptive
  • /test e2e-aws-fips
  • /test e2e-aws-image-registry
  • /test e2e-aws-jenkins
  • /test e2e-aws-multitenant
  • /test e2e-aws-ovn
  • /test e2e-aws-proxy
  • /test e2e-aws-serial
  • /test e2e-aws-upgrade
  • /test e2e-azure
  • /test e2e-gcp
  • /test e2e-gcp-builds
  • /test e2e-gcp-csi
  • /test e2e-gcp-disruptive
  • /test e2e-gcp-image-ecosystem
  • /test e2e-gcp-upgrade
  • /test e2e-metal-ipi
  • /test e2e-metal-ipi-ovn-dualstack
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-metal-ipi-virtualmedia
  • /test e2e-openstack
  • /test e2e-openstack-serial
  • /test e2e-vsphere
  • /test images
  • /test okd-e2e-gcp
  • /test verify
  • /test verify-deps
  • /test extended_gssapi
  • /test extended_ldap_groups
  • /test extended_networking

Use /test all to run the following jobs:

  • pull-ci-openshift-origin-master-e2e-agnostic-cmd
  • pull-ci-openshift-origin-master-e2e-aws-csi
  • pull-ci-openshift-origin-master-e2e-aws-disruptive
  • pull-ci-openshift-origin-master-e2e-aws-fips
  • pull-ci-openshift-origin-master-e2e-aws-serial
  • pull-ci-openshift-origin-master-e2e-gcp
  • pull-ci-openshift-origin-master-e2e-gcp-builds
  • pull-ci-openshift-origin-master-e2e-gcp-csi
  • pull-ci-openshift-origin-master-e2e-gcp-disruptive
  • pull-ci-openshift-origin-master-e2e-gcp-upgrade
  • pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-origin-master-images
  • pull-ci-openshift-origin-master-verify
  • pull-ci-openshift-origin-master-verify-deps
Details

In response to this:

/test pj-rehearse

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@andfasano

Copy link
Copy Markdown
Contributor Author

/test e2e-metal-ipi

@stbenjam

stbenjam commented Mar 2, 2021

Copy link
Copy Markdown
Member

/retest

@stbenjam

stbenjam commented Mar 2, 2021

Copy link
Copy Markdown
Member

/lgtm

@stbenjam

stbenjam commented Mar 9, 2021

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Mar 9, 2021
@stbenjam

Copy link
Copy Markdown
Member

Could we get another review please @deads2k?

Comment thread test/extended/baremetal/hosts.go Outdated
@openshift-ci-robot openshift-ci-robot removed the lgtm Indicates that a PR is ready to be merged. label Mar 15, 2021
@stbenjam

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Mar 15, 2021
@deads2k

deads2k commented Mar 15, 2021

Copy link
Copy Markdown
Contributor

/approve

@deads2k

deads2k commented Mar 15, 2021

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andfasano, deads2k, stbenjam

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 15, 2021
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

2 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-gcp-builds

@andfasano

Copy link
Copy Markdown
Contributor Author

/test e2e-gcp-builds
/test e2e-gcp-disruptive

@andfasano

Copy link
Copy Markdown
Contributor Author

/test e2e-gcp-builds

@andfasano

Copy link
Copy Markdown
Contributor Author

/test e2e-gcp
/test e2e-gcp-upgrade

@andfasano

Copy link
Copy Markdown
Contributor Author

/test e2e-gcp-upgrade

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-gcp-upgrade

I don't know if this is ever going to pass. It's been failing since 2021-03-15

@deads2k It seems unrelated to our PR, is someone looking at this or can you override it?

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-gcp-upgrade

@stbenjam

stbenjam commented Mar 19, 2021

Copy link
Copy Markdown
Member

@deads2k Could you help here? e2e-gcp-upgrade has succeeded twice now, but it takes so much time that prow says it wants to retest 8 jobs... and then e2e-gcp-upgrade might fail again, causing the cycle to repeat.

How does anything land in this repo? CI has been trying since 3/15 to land this PR.

@stbenjam

Copy link
Copy Markdown
Member

/retest

@stbenjam

Copy link
Copy Markdown
Member

/retest

sigh

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-aws-serial

@stbenjam

Copy link
Copy Markdown
Member

/retest

1 similar comment
@stbenjam

Copy link
Copy Markdown
Member

/retest

@openshift-merge-robot
openshift-merge-robot merged commit 2de32d9 into openshift:master Mar 21, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants