Skip to content

NodeName was being cleared for oc debug node/X instead of set - #22086

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
smarterclayton:fix_node
Feb 26, 2019
Merged

NodeName was being cleared for oc debug node/X instead of set#22086
openshift-merge-robot merged 1 commit into
openshift:masterfrom
smarterclayton:fix_node

Conversation

@smarterclayton

@smarterclayton smarterclayton commented Feb 19, 2019

Copy link
Copy Markdown
Contributor

Now that input can vary by output host, move setting this value into approximatePodForObject.

@openshift-ci-robot openshift-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Feb 19, 2019
@smarterclayton smarterclayton changed the title Create a kube-controller-manager image that contains ceph-common NodeName was being cleared for oc debug node/X instead of set Feb 19, 2019
@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 19, 2019
@smarterclayton

Copy link
Copy Markdown
Contributor Author

@sjenning

@sjenning

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Feb 19, 2019
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sjenning, smarterclayton

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

2 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

5 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

3 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@soltysh

soltysh commented Feb 21, 2019

Copy link
Copy Markdown
Contributor

/retest

1 similar comment
@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@jeremyeder

Copy link
Copy Markdown
Contributor

thank you lord

@soltysh

soltysh commented Feb 22, 2019

Copy link
Copy Markdown
Contributor

/retest

1 similar comment
@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

3 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@smarterclayton

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

3 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@soltysh

soltysh commented Feb 25, 2019

Copy link
Copy Markdown
Contributor

/retest

@smarterclayton

Copy link
Copy Markdown
Contributor Author

/refresh

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

1 similar comment
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants