Skip to content

Deprecate installer_checkpoint - #11196

Merged
mtnbikenc merged 1 commit into
openshift:devel-40from
mtnbikenc:rm-installer-checkpoint
Feb 25, 2019
Merged

Deprecate installer_checkpoint#11196
mtnbikenc merged 1 commit into
openshift:devel-40from
mtnbikenc:rm-installer-checkpoint

Conversation

@mtnbikenc

Copy link
Copy Markdown
Member

This role is no longer necessary.

This role is no longer necessary.
@openshift-ci-robot openshift-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Feb 14, 2019
@mtnbikenc

Copy link
Copy Markdown
Member Author

/retest

2 similar comments
@mtnbikenc

Copy link
Copy Markdown
Member Author

/retest

@mtnbikenc

Copy link
Copy Markdown
Member Author

/retest

@mtnbikenc

Copy link
Copy Markdown
Member Author
level=fatal msg="failed to initialize the cluster: timed out waiting for the condition"

/retest

@sdodson

sdodson commented Feb 22, 2019

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Feb 22, 2019
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mtnbikenc, sdodson

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot

openshift-ci-robot commented Feb 22, 2019

Copy link
Copy Markdown

@mtnbikenc: The following test failed for commit c71b444, say /retest to rerun them:

Test name Details Rerun command
ci/prow/e2e-aws-scaleup link /test e2e-aws-scaleup

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@mtnbikenc

Copy link
Copy Markdown
Member Author

The test passed, then flaked when it reran after lgtm. Ugh.

level=fatal msg="failed to fetch Cluster: failed to generate asset \"Cluster\": failed to create cluster: failed to apply using Terraform"

/retest

@mtnbikenc

Copy link
Copy Markdown
Member Author
fatal: [localhost]: FAILED! => 
  msg: 'The conditional check ''svc.results.results[0].status.loadBalancer.ingress | length > 0'' failed. The error was: error while evaluating conditional

/retest

@mtnbikenc

Copy link
Copy Markdown
Member Author

/refresh

@mtnbikenc

Copy link
Copy Markdown
Member Author
Failing tests:

[Conformance][Area:Networking][Feature:Router] The default ClusterIngress should support default wildcard reencrypt routes through external DNS [Suite:openshift/conformance/parallel/minimal]
[Feature:Builds] build have source revision metadata  started build should contain source revision information [Suite:openshift/conformance/parallel]
[Feature:Builds] result image should have proper labels set  S2I build from a template should create a image from "test-s2i-build.json" template with proper Docker labels [Suite:openshift/conformance/parallel]
[sig-apps] StatefulSet [k8s.io] Basic StatefulSet functionality [StatefulSetBasic] should perform rolling updates and roll backs of template modifications with PVCs [Suite:openshift/conformance/parallel] [Suite:k8s]

/retest

@vrutkovs

Copy link
Copy Markdown
Contributor
Failing tests:

[Feature:Platform] Managed cluster should have no crashlooping pods in core namespaces over two minutes [Suite:openshift/conformance/parallel]

/retest

@sdodson

sdodson commented Feb 25, 2019

Copy link
Copy Markdown
Member

/retest

@mtnbikenc

Copy link
Copy Markdown
Member Author

#326

Failing tests:

[Feature:Platform] Managed cluster should have no crashlooping pods in core namespaces over two minutes [Suite:openshift/conformance/parallel]

@mtnbikenc

mtnbikenc commented Feb 25, 2019

Copy link
Copy Markdown
Member Author

Crashlooping pods is fixed by openshift/machine-config-operator#495

@mtnbikenc

Copy link
Copy Markdown
Member Author

Two consecutive failures on the same test with a known fix in the works. Merging.

@mtnbikenc
mtnbikenc merged commit 8d24ee7 into openshift:devel-40 Feb 25, 2019
@mtnbikenc
mtnbikenc deleted the rm-installer-checkpoint branch February 25, 2019 21:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants