Skip to content

Bug 1706689: pkg: Catch up to the api change introduced in openshift/api#308 - #715

Merged
imcleod merged 2 commits into
openshift:masterfrom
abhinavdahiya:infra_api_change
May 8, 2019
Merged

Bug 1706689: pkg: Catch up to the api change introduced in openshift/api#308 #715
imcleod merged 2 commits into
openshift:masterfrom
abhinavdahiya:infra_api_change

Conversation

@abhinavdahiya

Copy link
Copy Markdown
Contributor

Brins in changes from openshift/api#308

```console
$ dep version
dep:
 version     : v0.5.0
 build date  : 2018-07-26
 git hash    : 224a564
 go version  : go1.10.3
 go compiler : gc
 platform    : linux/amd64
 features    : ImportDuringSolve=false
$ dep ensure -update github.com/openshift/api github.com/openshift/client-go
```
@openshift-ci-robot
openshift-ci-robot requested a review from runcom May 7, 2019 14:12
@openshift-ci-robot openshift-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label May 7, 2019
@runcom runcom changed the title pkg: Catch up to the api change introduced in openshift/api#308 Bug 1706689: pkg: Catch up to the api change introduced in openshift/api#308 May 7, 2019
@kikisdeliveryservice

kikisdeliveryservice commented May 7, 2019

Copy link
Copy Markdown
Contributor

since this requires openshift/installer#1718 to be merged first, for clarity:

/hold

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 7, 2019
@abhinavdahiya

Copy link
Copy Markdown
Contributor Author

/retest

openshift/installer#1718 merged
/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 8, 2019
@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label May 8, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abhinavdahiya, runcom

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 8, 2019
@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

2 similar comments
@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

can you rebase to pick can you rebase to pick #719

@abhinavdahiya

Copy link
Copy Markdown
Contributor Author

can you rebase to pick can you rebase to pick #719

you don't need to rebase for that, i can delete the project in ci cluster if this run fails before running a retest to bring in updated MCO master.

@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

this is good to go, but I'm holding just for the beta

/hold

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 8, 2019
@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 8, 2019
@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

/hold cancel

@imcleod

imcleod commented May 8, 2019

Copy link
Copy Markdown
Contributor

/retest

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

This should be in the merge pool, though it can take a while for it to go in.

@runcom

runcom commented May 8, 2019

Copy link
Copy Markdown
Member

/refresh

@imcleod
imcleod merged commit 6e7615a into openshift:master May 8, 2019
@abhinavdahiya
abhinavdahiya deleted the infra_api_change branch May 8, 2019 18:14
ptalgulk01 pushed a commit to ptalgulk01/machine-config-operator that referenced this pull request May 15, 2026
Bug 1706689: pkg: Catch up to the api change introduced in openshift/api#308
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants