Skip to content

pkg/operator: use fake clients in unit - #452

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
runcom:use-fakes-units
Feb 20, 2019
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
runcom:use-fakes-units

Conversation

@runcom

@runcom runcom commented Feb 18, 2019

Copy link
Copy Markdown
Member

Signed-off-by: Antonio Murdaca runcom@linux.com

- What I did

I added a wrapper around the ClusterOperators client mainly for unit tests in #442 which turns out to be totally unnecessary since I've learned about fake clients in kube land. This PR removes that wrapper and turns the code back to its state and just use plain fake clients in unit tests which will reduce code and not needed abstractions.

- How to verify it

- Description for the changelog

Signed-off-by: Antonio Murdaca <runcom@linux.com>
@openshift-ci-robot openshift-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Feb 18, 2019
@cgwalters

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Feb 19, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cgwalters, runcom

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@LorbusChris

Copy link
Copy Markdown
Contributor

/test e2e-aws

@runcom

runcom commented Feb 19, 2019

Copy link
Copy Markdown
Member Author

/retest

@runcom

runcom commented Feb 19, 2019

Copy link
Copy Markdown
Member Author

aws limit

/retest

@ashcrow

ashcrow commented Feb 19, 2019

Copy link
Copy Markdown
Member

Bootstrap user flake

/retest

@ashcrow

ashcrow commented Feb 19, 2019

Copy link
Copy Markdown
Member

AWS is unhappy.

aws_instance.master.0: InvalidInstanceID.NotFound:

@runcom

runcom commented Feb 19, 2019

Copy link
Copy Markdown
Member Author

/retest

@runcom

runcom commented Feb 19, 2019

Copy link
Copy Markdown
Member Author

unit tests fixed in #457

/retest

@openshift-merge-robot
openshift-merge-robot merged commit 4440b87 into openshift:master Feb 20, 2019
@runcom
runcom deleted the use-fakes-units branch February 20, 2019 00:59
ptalgulk01 pushed a commit to ptalgulk01/machine-config-operator that referenced this pull request May 15, 2026
pkg/operator: use fake clients in unit
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants