kubeletconfig: fix updates, add unit test, race condition fix - #437
openshift-merge-robot merged 2 commits into
Conversation
|
/approve /cc @sjenning |
|
odd... failure in a different test. |
|
That's known #417 |
|
/retest I'll take a look at the flaky test |
18fbafe to
f36769d
Compare
|
@rphillips thanks for the flake test fixes ❤️ could you add "close #417" to the first comment as well? |
|
|
f36769d to
57ff65d
Compare
|
/retest |
|
/retest |
3 similar comments
|
/retest |
|
/retest |
|
/retest |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: rphillips, runcom The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
1 similar comment
|
/retest Please review the full test history for this PR and help us cut down flakes. |
|
/retest |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
1 similar comment
|
/retest Please review the full test history for this PR and help us cut down flakes. |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
…kubelet_configs kubeletconfig: fix updates, add unit test, race condition fix
- What I did
@umohnani8 found a bug in kubeletconfig updates. This PR adds a unit test and fixes the underlying validation.
- How to verify it
Unit test
- Description for the changelog
Fix KubeletConfig updates not being applied correctly to the MachineConfig.
Also fixes race conditions found in:
Close #417