Skip to content

test CI for no code - #2732

Closed
miheer wants to merge 1 commit into
openshift:masterfrom
miheer:test
Closed

miheer wants to merge 1 commit into
openshift:masterfrom
miheer:test

Conversation

@miheer

@miheer miheer commented Aug 25, 2021

Copy link
Copy Markdown
Contributor

- What I did

- How to verify it

- Description for the changelog

@openshift-ci

openshift-ci Bot commented Aug 25, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: miheer

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Aug 25, 2021

Copy link
Copy Markdown
Contributor

@miheer: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Rerun command
ci/prow/e2e-aws-techpreview-featuregate 0faf7d3 link /test e2e-aws-techpreview-featuregate
ci/prow/e2e-aws-single-node 0faf7d3 link /test e2e-aws-single-node

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

closing since this seems to be unused.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants