Skip to content

[release-4.5] Bug 1845636: etcd-quorum-guard: update containers section in deployment.yaml - #1819

Merged
openshift-merge-robot merged 1 commit into
openshift:release-4.5from
kikisdeliveryservice:fix-quorum-guard-4.5
Jun 18, 2020
Merged

openshift-merge-robot merged 1 commit into
openshift:release-4.5from
kikisdeliveryservice:fix-quorum-guard-4.5

Conversation

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

Made the following changes as settings were not getting
picked up:

  • move and correct initialDelaySeconds and periodSeconds
  • reorder container sub-elements to nest image under name
  • add failureThreshold and timeoutSeconds

Closes: BZ 1845636

@openshift-ci-robot openshift-ci-robot added bugzilla/severity-medium Referenced Bugzilla bug's severity is medium for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. labels Jun 12, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@kikisdeliveryservice: This pull request references Bugzilla bug 1845636, which is valid. The bug has been updated to refer to the pull request using the external bug tracker.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target release (4.5.0) matches configured target release for branch (4.5.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, ON_DEV, POST, POST)
  • dependent bug Bugzilla bug 1829923 is in the state MODIFIED, which is one of the valid states (MODIFIED, ON_QA, VERIFIED)
  • dependent Bugzilla bug 1829923 targets the "4.6.0" release, which is one of the valid target releases: 4.6.0, 4.6.z
  • bug has dependents
Details

In response to this:

Bug 1845636: etcd-quorum-guard: update containers section in deployment.yaml

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 12, 2020
@kikisdeliveryservice
kikisdeliveryservice requested review from hexfusion, ironcladlou and wking and removed request for ashcrow and ericavonb June 12, 2020 02:28
@kikisdeliveryservice

Copy link
Copy Markdown
Contributor Author

one mistake

/hold

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jun 12, 2020
Made the following changes as settings were not getting
picked up:
- move and correct initialDelaySeconds and periodSeconds
- reorder container sub-elements to nest image under name
- add failureThreshold and timeoutSeconds

Closes: BZ 1845636
@kikisdeliveryservice kikisdeliveryservice changed the title Bug 1845636: etcd-quorum-guard: update containers section in deployment.yaml [release-4.5] Bug 1845636: etcd-quorum-guard: update containers section in deployment.yaml Jun 12, 2020
@kikisdeliveryservice

Copy link
Copy Markdown
Contributor Author

ok should be good

/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jun 12, 2020
@hexfusion

Copy link
Copy Markdown
Contributor

/retest

2 similar comments
@ironcladlou

Copy link
Copy Markdown
Contributor

/retest

@ashcrow

ashcrow commented Jun 15, 2020

Copy link
Copy Markdown
Member

/retest

@kikisdeliveryservice

kikisdeliveryservice commented Jun 15, 2020

Copy link
Copy Markdown
Contributor Author

this test is broken
/skip e2e-aws-disruptive

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor Author

/override ci/prow/e2e-aws-disruptive

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@kikisdeliveryservice: Overrode contexts on behalf of kikisdeliveryservice: ci/prow/e2e-aws-disruptive

Details

In response to this:

/override ci/prow/e2e-aws-disruptive

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@hexfusion

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Jun 17, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hexfusion, kikisdeliveryservice

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [kikisdeliveryservice]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor Author

/assign @ashcrow

@ashcrow

ashcrow commented Jun 17, 2020

Copy link
Copy Markdown
Member

Backport of #1818

@ashcrow ashcrow added the staff-eng-approved Indicates a release branch PR has been approved by a staff engineer (formerly group/pillar lead). label Jun 17, 2020
@ashcrow

ashcrow commented Jun 17, 2020

Copy link
Copy Markdown
Member

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit 7b1eb84 into openshift:release-4.5 Jun 18, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@kikisdeliveryservice: All pull requests linked via external trackers have merged: openshift/machine-config-operator#1819. Bugzilla bug 1845636 has been moved to the MODIFIED state.

Details

In response to this:

[release-4.5] Bug 1845636: etcd-quorum-guard: update containers section in deployment.yaml

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. bugzilla/severity-medium Referenced Bugzilla bug's severity is medium for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. lgtm Indicates that a PR is ready to be merged. staff-eng-approved Indicates a release branch PR has been approved by a staff engineer (formerly group/pillar lead).

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants