Skip to content

[release-4.4] Bug 1843928: gcp-routes: move to MCO, implement downfile, tweak timing - #1780

Merged
openshift-merge-robot merged 3 commits into
openshift:release-4.4from
openshift-cherrypick-robot:cherry-pick-1670-to-release-4.4
Jun 11, 2020
Merged

[release-4.4] Bug 1843928: gcp-routes: move to MCO, implement downfile, tweak timing#1780
openshift-merge-robot merged 3 commits into
openshift:release-4.4from
openshift-cherrypick-robot:cherry-pick-1670-to-release-4.4

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #1670

/assign sttts

squeed added 3 commits June 4, 2020 12:35
They're missing a lot of things, but let's just add a reasonable Infra
object for now.
A new mode for gcp-routes allows for marking individual VIPs as down. If
that's available, use it.
OpenShift wants a pretty different version of gcp-routes.service
compared to the generic case. So we really should just control it
ourselves.

This creates a service, openshift-gcp-routes.service, that conflicts
with the gcp-routes.service from the RHCOS overlay. It also picks up
some pending improvements to it, namely downfile support.
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: Bugzilla bug 1802534 has been cloned as Bugzilla bug 1843928. Retitling PR to link against new bug.
/retitle [release-4.4] Bug 1843928: gcp-routes: move to MCO, implement downfile, tweak timing

Details

In response to this:

[release-4.4] Bug 1802534: gcp-routes: move to MCO, implement downfile, tweak timing

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot changed the title [release-4.4] Bug 1802534: gcp-routes: move to MCO, implement downfile, tweak timing [release-4.4] Bug 1843928: gcp-routes: move to MCO, implement downfile, tweak timing Jun 4, 2020
@openshift-ci-robot openshift-ci-robot added the bugzilla/severity-high Referenced Bugzilla bug's severity is high for the branch this PR is targeting. label Jun 4, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: This pull request references Bugzilla bug 1843928, which is invalid:

  • expected dependent Bugzilla bug 1802534 to be in one of the following states: VERIFIED, RELEASE_PENDING, CLOSED (ERRATA), but it is ON_QA instead

Comment /bugzilla refresh to re-evaluate validity if changes to the Bugzilla bug are made, or edit the title of this pull request to link to a different bug.

Details

In response to this:

[release-4.4] Bug 1843928: gcp-routes: move to MCO, implement downfile, tweak timing

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the bugzilla/invalid-bug Indicates that a referenced Bugzilla bug is invalid for the branch this PR is targeting. label Jun 4, 2020
@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 4, 2020
@p0lyn0mial

Copy link
Copy Markdown

/test e2e-gcp-upgrade

1 similar comment
@sdodson

sdodson commented Jun 4, 2020

Copy link
Copy Markdown
Member

/test e2e-gcp-upgrade

@sinnykumari

Copy link
Copy Markdown
Contributor

/retest

1 similar comment
@sttts

sttts commented Jun 5, 2020

Copy link
Copy Markdown
Contributor

/retest

@vrutkovs

vrutkovs commented Jun 8, 2020

Copy link
Copy Markdown
Contributor

/retest
/refresh

@wking

wking commented Jun 10, 2020

Copy link
Copy Markdown
Member

/bugzilla refresh

@openshift-ci-robot openshift-ci-robot added the bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. label Jun 10, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wking: This pull request references Bugzilla bug 1843928, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target release (4.4.z) matches configured target release for branch (4.4.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, ON_DEV, POST, POST)
  • dependent bug Bugzilla bug 1802534 is in the state VERIFIED, which is one of the valid states (VERIFIED, RELEASE_PENDING, CLOSED (ERRATA))
  • dependent Bugzilla bug 1802534 targets the "4.5.0" release, which is one of the valid target releases: 4.5.0, 4.5.z
  • bug has dependents
Details

In response to this:

/bugzilla refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot removed the bugzilla/invalid-bug Indicates that a referenced Bugzilla bug is invalid for the branch this PR is targeting. label Jun 10, 2020
@sttts

sttts commented Jun 10, 2020

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Jun 10, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ashcrow, openshift-cherrypick-robot, sttts

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knobunc knobunc added the cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. label Jun 10, 2020
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

1 similar comment
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit b6c95fe into openshift:release-4.4 Jun 11, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: All pull requests linked via external trackers have merged: openshift/machine-config-operator#1780. Bugzilla bug 1843928 has been moved to the MODIFIED state.

Details

In response to this:

[release-4.4] Bug 1843928: gcp-routes: move to MCO, implement downfile, tweak timing

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. bugzilla/severity-high Referenced Bugzilla bug's severity is high for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.