Skip to content

[release-4.2] Bug 1801826: kubelet: add more kube reservation to protect node - #1459

Closed
rphillips wants to merge 1 commit into
openshift:release-4.2from
rphillips:fixes/add_kube_reservation_4.2
Closed

rphillips wants to merge 1 commit into
openshift:release-4.2from
rphillips:fixes/add_kube_reservation_4.2

Conversation

@rphillips

Copy link
Copy Markdown
Contributor

Cherry pick of: #1450

bump Kubelet reservations

- What I did

- How to verify it

- Description for the changelog

includes: kubelet, container runtime
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@rphillips: This pull request references Bugzilla bug 1801826, which is invalid:

  • expected dependent Bugzilla bug 1800319 to be in one of the following states: VERIFIED, RELEASE_PENDING, CLOSED (ERRATA), but it is POST instead
  • expected dependent Bugzilla bug 1801824 to be in one of the following states: VERIFIED, RELEASE_PENDING, CLOSED (ERRATA), but it is NEW instead

Comment /bugzilla refresh to re-evaluate validity if changes to the Bugzilla bug are made, or edit the title of this pull request to link to a different bug.

Details

In response to this:

[release-4.2] Bug 1801826: kubelet: add more kube reservation to protect node

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the bugzilla/invalid-bug Indicates that a referenced Bugzilla bug is invalid for the branch this PR is targeting. label Feb 11, 2020
@rphillips
rphillips requested a review from sjenning February 11, 2020 17:11
@openshift-ci-robot openshift-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Feb 11, 2020
@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

/approve

@kikisdeliveryservice

Copy link
Copy Markdown
Contributor

/assign @sjenning @mrunalp

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: kikisdeliveryservice, rphillips

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [kikisdeliveryservice]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 11, 2020
@rphillips

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 14, 2020
@rphillips rphillips closed this Feb 25, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. bugzilla/invalid-bug Indicates that a referenced Bugzilla bug is invalid for the branch this PR is targeting. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants