Skip to content

Bug 1878163: Build with golang 1.15 - #348

Merged
openshift-merge-robot merged 3 commits into
openshift:masterfrom
marun:golang-1.15
Sep 18, 2020
Merged

Bug 1878163: Build with golang 1.15#348
openshift-merge-robot merged 3 commits into
openshift:masterfrom
marun:golang-1.15

Conversation

@marun

@marun marun commented Sep 10, 2020

Copy link
Copy Markdown

No description provided.

@marun marun changed the title Build with golang 1.15 Bug 1878163: Build with golang 1.15 Sep 11, 2020
@openshift-ci-robot openshift-ci-robot added the bugzilla/severity-high Referenced Bugzilla bug's severity is high for the branch this PR is targeting. label Sep 11, 2020
@openshift-ci-robot

Copy link
Copy Markdown

@marun: This pull request references Bugzilla bug 1878163, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target release (4.6.0) matches configured target release for branch (4.6.0)
  • bug is in the state NEW, which is one of the valid states (NEW, ASSIGNED, ON_DEV, POST, POST)
Details

In response to this:

Bug 1878163: Build with golang 1.15

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. label Sep 11, 2020
@marun

marun commented Sep 14, 2020

Copy link
Copy Markdown
Author

/retest

@marun

marun commented Sep 14, 2020

Copy link
Copy Markdown
Author

This branch is now building with golang 1.15 so ci should be capable of passing.

@marun

marun commented Sep 15, 2020

Copy link
Copy Markdown
Author

/retest

3 similar comments
@marun

marun commented Sep 15, 2020

Copy link
Copy Markdown
Author

/retest

@marun

marun commented Sep 16, 2020

Copy link
Copy Markdown
Author

/retest

@marun

marun commented Sep 16, 2020

Copy link
Copy Markdown
Author

/retest

@yselkowitz

Copy link
Copy Markdown

FYI there is no golang-1.15 for RHEL 7 yet, only for RHEL 8. As the RPM is built for both, this needs to be kept in mind.

@marun

marun commented Sep 16, 2020

Copy link
Copy Markdown
Author

/retest

@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

/retest

@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

/test unit

@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

Unit tests are chronically racey.

@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

I can't replicate the verify failure locally, ocp wins again. Given that this was working up until this week, maybe the ci migration from 3.x to 4.x is to blame?

@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

/retest

2 similar comments
@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

/retest

@marun

marun commented Sep 17, 2020

Copy link
Copy Markdown
Author

/retest

…om source

Remove installation from source of recent bash and protoc 3.0.0 now
that both are available on the ci build root image.

Squash with UPSTREAM: <carry>: Enable build, test and verify
@marun

marun commented Sep 18, 2020

Copy link
Copy Markdown
Author

/retest

@marun
marun requested a review from tnozicka September 18, 2020 05:06
@marun

marun commented Sep 18, 2020

Copy link
Copy Markdown
Author

/retest

@openshift-ci-robot

Copy link
Copy Markdown

@marun: The following test failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/prow/k8s-e2e-gcp-serial 583e078 link /test k8s-e2e-gcp-serial

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@sttts

sttts commented Sep 18, 2020

Copy link
Copy Markdown

/lgtm
/approve

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 18, 2020
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: marun, sttts

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 18, 2020
@openshift-merge-robot
openshift-merge-robot merged commit 7f9e863 into openshift:master Sep 18, 2020
@openshift-ci-robot

Copy link
Copy Markdown

@marun: Some pull requests linked via external trackers have merged:

The following pull requests linked via external trackers have not merged:

These pull request must merge or be unlinked from the Bugzilla bug in order for it to move to the next state.

Bugzilla bug 1878163 has not been moved to the MODIFIED state.

Details

In response to this:

Bug 1878163: Build with golang 1.15

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. bugzilla/severity-high Referenced Bugzilla bug's severity is high for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants