Skip to content

Baremetal: Generate CR to configure baremetal. - #2958

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
imain:generate_cr
Jan 23, 2020
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
imain:generate_cr

Conversation

@imain

@imain imain commented Jan 21, 2020

Copy link
Copy Markdown
Contributor

Use a template to generate the baremetal config CR for use by the MAO
and the baremetal operator.

@openshift-ci-robot openshift-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Jan 21, 2020
@stbenjam

stbenjam commented Jan 21, 2020

Copy link
Copy Markdown
Member

/cc @sadasu Can you have a look at this?

Any chance we could get a dev-scripts PR put together that pulls down this PR and MAO (openshift/machine-api-operator#470) and tests them together, without the configmap hacks?

Use a template to generate the baremetal config CR for use by the MAO
and the baremetal operator.
@stbenjam

Copy link
Copy Markdown
Member

/label platform/baremetal

@openshift-ci-robot openshift-ci-robot added the platform/baremetal IPI bare metal hosts platform label Jan 22, 2020
@metal3ci

Copy link
Copy Markdown

Build SUCCESS, see build http://10.8.144.11:8080/job/dev-tools/1437/

@stbenjam

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Jan 23, 2020
@stbenjam

Copy link
Copy Markdown
Member

/retest

@stbenjam

Copy link
Copy Markdown
Member

/assign @sdodson

This is part of the implementation for openshift/enhancements#119. The CRD was added to the MAO in openshift/machine-api-operator#460.

We now need the installer to write out a CR with values from our installation configuration. Currently, MAO reads this by a manually created config map, the goal here is to remove that entirely.

@sdodson

sdodson commented Jan 23, 2020

Copy link
Copy Markdown
Member

/approve

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sdodson

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 23, 2020
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

2 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@stbenjam

Copy link
Copy Markdown
Member

/test e2e-aws-upgrade

@imain

imain commented Jan 23, 2020

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

1 similar comment
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-ci-robot

openshift-ci-robot commented Jan 23, 2020

Copy link
Copy Markdown
Contributor

@imain: The following tests failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/prow/e2e-aws-scaleup-rhel7 9dfb595 link /test e2e-aws-scaleup-rhel7
ci/prow/e2e-aws-fips 9dfb595 link /test e2e-aws-fips
ci/prow/e2e-libvirt 9dfb595 link /test e2e-libvirt

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. platform/baremetal IPI bare metal hosts platform size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants