Skip to content

docs/user/versioning: Drop obsolete versioning claims - #2442

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:drop-sem-ver-claims
Oct 2, 2019
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:drop-sem-ver-claims

Conversation

@wking

@wking wking commented Oct 1, 2019

Copy link
Copy Markdown
Member

I'd added these in 28592d0 (#935). But v0.16.1 was the last installer release. Since then, OpenShift tooling picks whatever the most recent commit is in the appropriate release-* branch when building releases, so we don't have a way to set the SemVer versions on the command-line API that the removed document depends on. We will attempt to preserve any command-line API that occurs in official documentation, but we can no longer make formal SemVer statements about breaking compat.

I'd added these in 28592d0 (docs/user/versioning: Document APIs
covered by our Semantic Versioning, 2018-12-17, openshift#935).  But v0.16.1
was the last installer release.  Since then, OpenShift tooling picks
whatever the most recent commit is in the appropriate release-* branch
when building releases, so we don't have a way to set the SemVer
versions on the command-line API that the removed document depends on.
We will attempt to preserve any command-line API that occurs in
official documentation, but we can no longer make formal SemVer
statements about breaking compat.
@openshift-ci-robot openshift-ci-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Oct 1, 2019
@abhinavdahiya

Copy link
Copy Markdown
Contributor

/approve

also for review
/cc @sdodson

@sdodson

sdodson commented Oct 2, 2019

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 2, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abhinavdahiya, sdodson, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [abhinavdahiya,sdodson,wking]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit 8d8beb5 into openshift:master Oct 2, 2019
@wking
wking deleted the drop-sem-ver-claims branch October 2, 2019 18:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants