Skip to content

Bug 1706689: asset/manifests: update infrastructure to set public apiserver url - #1727

Merged
openshift-merge-robot merged 4 commits into
openshift:masterfrom
abhinavdahiya:infra_api_changes_public
May 8, 2019
Merged

Bug 1706689: asset/manifests: update infrastructure to set public apiserver url#1727
openshift-merge-robot merged 4 commits into
openshift:masterfrom
abhinavdahiya:infra_api_changes_public

Conversation

@abhinavdahiya

Copy link
Copy Markdown
Contributor

PR 1 kept the apiServerURL field unchanged ie set to internal url for Kubernetes API so that
the consumers of the field can move away from it. With KAS-O 2 change merged, we can move forward

Builds on #1718 and Requires openshift/cluster-kube-apiserver-operator#465

Brins in changes from openshift/api#308

```console
$ dep version
dep:
 version     : v0.5.0
 build date  : 2018-07-26
 git hash    : 224a564
 go version  : go1.10.3
 go compiler : gc
 platform    : linux/amd64
 features    : ImportDuringSolve=false
$ dep ensure -update github.com/openshift/api github.com/openshift/client-go
```
Catch up to the api change introduced in [1]

PR [1] introduces breaking change for `.status.apiServerURL` therefore, this only adds the new
field `.status.apiServerInternalURL` so that consumers like KAS-O and MCO can move to this new field and
then installer can move the `.status.apiServerURL` to public url for Kubernetes API.

[1]: openshift/api#308
PR [1] dropped the Failing condition and CVO moved to still reports Failing by using a local definiton [2].

This defines a local definition for failing condition like in CVO.

[1]: openshift/api@a9fb3b1
[2]: openshift/cluster-version-operator@0b47fc9
PR [1] kept the `apiServerURL` field unchanged ie set to internal url for Kubernetes API so that
the consumers of the field can move away from it. With KAS-O [2] change merged, we can move forward

[1]: openshift#1718
[2]: openshift/cluster-kube-apiserver-operator#465
@openshift-ci-robot openshift-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels May 8, 2019
@wking

wking commented May 8, 2019

Copy link
Copy Markdown
Member

/lgtm
/hold

@openshift-ci-robot openshift-ci-robot added do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. lgtm Indicates that a PR is ready to be merged. labels May 8, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abhinavdahiya, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [abhinavdahiya,wking]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sdodson

sdodson commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

@abhinavdahiya

Copy link
Copy Markdown
Contributor Author

#1730 merged
/retest
/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 8, 2019
@deads2k

deads2k commented May 8, 2019

Copy link
Copy Markdown
Contributor

1730 merged

/retest

@deads2k

deads2k commented May 8, 2019

Copy link
Copy Markdown
Contributor

/retest

@wking

wking commented May 8, 2019

Copy link
Copy Markdown
Member

#1730 landed.

/retest

@deads2k

deads2k commented May 8, 2019

Copy link
Copy Markdown
Contributor

/retest

@eparis

eparis commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

1 similar comment
@eparis

eparis commented May 8, 2019

Copy link
Copy Markdown
Member

/retest

@openshift openshift deleted a comment from deads2k May 8, 2019
@timlnx

timlnx commented May 8, 2019

Copy link
Copy Markdown

/retest

@wking

wking commented May 8, 2019

Copy link
Copy Markdown
Member

I'm not clear on why AWS is getting retested now. We were green here and here, and at least the e2e-aws job kicked off after the last master bump:

$ git log --first-parent --format='%ad %h %d %s' --date=iso -2 origin/master
2019-05-08 19:56:40 +0200 c91435c84  (origin/release-4.2, origin/release-4.1, origin/master, origin/HEAD) Merge pull request #1730 from abhinavdahiya/upi_image_fix
2019-05-08 05:06:26 +0200 23aac5288  Merge pull request #1718 from abhinavdahiya/infra_api_changes
$ date --iso=m --utc --date '2019-05-08 19:56:40 +0200'
2019-05-08T17:56+0000
$ date --iso=m --utc --date "@$(curl -s https://storage.googleapis.com/origin-ci-test/pr-logs/pull/openshift_installer/1727/pull-ci-openshift-installer-master-e2e-aws/5738/started.json | jq -r .timestamp)"
2019-05-08T18:06+0000

Anyhow, with that UPI image the only bump, I'm ok force-merging this to unstick the beta release.

@wking

wking commented May 8, 2019

Copy link
Copy Markdown
Member

Ah, maybe a delayed Prow trigger:

Screenshot from 2019-05-08 12-58-31

@trown

trown commented May 8, 2019

Copy link
Copy Markdown

/test e2e-openstack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@abhinavdahiya: The following test failed, say /retest to rerun them all:

Test name Commit Details Rerun command
ci/prow/e2e-openstack 5987f7e link /test e2e-openstack

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot
openshift-merge-robot merged commit 3b6832c into openshift:master May 8, 2019
@abhinavdahiya
abhinavdahiya deleted the infra_api_changes_public branch May 8, 2019 21:55
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

9 participants