Skip to content

machines/libvirt: Drop volume paths - #1628

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
zeenix:drop-volume-paths
May 31, 2019
Merged

machines/libvirt: Drop volume paths#1628
openshift-merge-robot merged 1 commit into
openshift:masterfrom
zeenix:drop-volume-paths

Conversation

@zeenix

@zeenix zeenix commented Apr 16, 2019

Copy link
Copy Markdown
Contributor

Volume should be identified by their names and their paths should be left to their container pools. This change is needed to allow for custom storage locations.

This is the needed update in Installer for the actual change in the libvirt actuator: openshift/cluster-api-provider-libvirt#144

Fixes #308.

Volume should be identified by their names and their paths should be
left to their container pools. This change is needed to allow for custom
storage locations.

This is the needed update in Installer for the actual change in the
libvirt actuator: openshift/cluster-api-provider-libvirt#144
@openshift-ci-robot openshift-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Apr 16, 2019
@abhinavdahiya abhinavdahiya changed the title libvirt: Drop volume paths machines/libvirt: Drop volume paths Apr 16, 2019
@cgwalters

Copy link
Copy Markdown
Member

I tested this, and it
/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Apr 16, 2019
@zeenix

zeenix commented Apr 17, 2019

Copy link
Copy Markdown
Contributor Author

@cgwalters Thanks. It shouldn't work with existing actuator btw so I'm guessing you tested with an override for the actuator including openshift/cluster-api-provider-libvirt#144 ?

@cgwalters

Copy link
Copy Markdown
Member

It shouldn't work with existing actuator btw

Upon some inspection, turns out I'd recently changed my installer dev flow and while I'd built the installer from your PR I hadn't actually moved it into my $PATH...sorry for the noise.

@praveenkumar

Copy link
Copy Markdown
Contributor

/test e2e-libvirt

@zeenix

zeenix commented Apr 23, 2019

Copy link
Copy Markdown
Contributor Author

@praveenkumar The test will fail until openshift/cluster-api-provider-libvirt#144 is merged. The change there is backwards compatible but not here so we need to handle that PR first.

@zeenix

zeenix commented Apr 25, 2019

Copy link
Copy Markdown
Contributor Author

/test e2e-libvirt

1 similar comment
@zeenix

zeenix commented May 2, 2019

Copy link
Copy Markdown
Contributor Author

/test e2e-libvirt

@zeenix

zeenix commented May 20, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@zeenix

zeenix commented May 21, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@zeenix

zeenix commented May 21, 2019

Copy link
Copy Markdown
Contributor Author

/test e2e-libvirt

@praveenkumar

Copy link
Copy Markdown
Contributor

/retest

@zeenix

zeenix commented May 22, 2019

Copy link
Copy Markdown
Contributor Author

/test e2e-libvirt

1 similar comment
@zeenix

zeenix commented May 27, 2019

Copy link
Copy Markdown
Contributor Author

/test e2e-libvirt

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cgwalters, zeenix

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 31, 2019
@praveenkumar

Copy link
Copy Markdown
Contributor

/approved

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

1 similar comment
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit e3ba84d into openshift:master May 31, 2019
@zeenix
zeenix deleted the drop-volume-paths branch June 12, 2019 14:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

libvirt: can't create workers with non-default storage path

6 participants