Skip to content

Rename Go module from openshift.io/hypershift to github.com/openshift/hypershift - #72

Merged
openshift-merge-robot merged 1 commit into
openshift:mainfrom
ironcladlou:rename-module
Mar 1, 2021
Merged

openshift-merge-robot merged 1 commit into
openshift:mainfrom
ironcladlou:rename-module

Conversation

@ironcladlou

Copy link
Copy Markdown
Contributor

We've decided a vanity domain isn't justified. Rename the Go module to match the
VCS location for simplicity.

…/hypershift

We've decided a vanity domain isn't justified. Rename the Go module to match the
VCS location for simplicity.
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ironcladlou

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 1, 2021
@ironcladlou

Copy link
Copy Markdown
Contributor Author

/test e2e

@openshift-ci-robot

Copy link
Copy Markdown

@ironcladlou: The specified target(s) for /test were not found.
The following commands are available to trigger jobs:

  • /test e2e-aws
  • /test images
  • /test unit

Use /test all to run the following jobs:

  • pull-ci-openshift-hypershift-main-images
  • pull-ci-openshift-hypershift-main-unit
Details

In response to this:

/test e2e

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@ironcladlou

Copy link
Copy Markdown
Contributor Author

/test e2e-aws

@sjenning

sjenning commented Mar 1, 2021

Copy link
Copy Markdown
Contributor

waiting for e2e, but lgtm

@csrwng

csrwng commented Mar 1, 2021

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Mar 1, 2021
@openshift-merge-robot
openshift-merge-robot merged commit 5c52cdc into openshift:main Mar 1, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants